)]}'
{"id":"openstack%2Fpuppet-ovn~864030","triplet_id":"openstack%2Fpuppet-ovn~stable%2Fwallaby~Ia3112b9a219b90af718c49776da49078449f09a2","project":"openstack/puppet-ovn","branch":"stable/wallaby","topic":"ovn-ofctrl-wait-before-clear","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-11-16 11:33:51.000000000","reason":"Change was submitted"},"14985":{"account":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"last_update":"2022-11-16 11:33:51.000000000","reason":"Change was submitted"},"5756":{"account":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"last_update":"2022-11-16 11:33:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia3112b9a219b90af718c49776da49078449f09a2","subject":"Add support for ovn-ofctrl-wait-before-clear","status":"MERGED","created":"2022-11-08 16:10:51.000000000","updated":"2022-11-16 11:34:56.000000000","submitted":"2022-11-16 11:33:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864030-ovn-ofctrl-wait-before-clear","cherry_pick_of_change":863809,"cherry_pick_of_patch_set":7,"meta_rev_id":"cb5f351e7760950ade6501c61c45ee35d67ebc02","_number":864030,"virtual_id_number":864030,"owner":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-16 11:33:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-11-14 15:16:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-11-16 11:12:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-11-16 11:12:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-08 16:32:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-14 15:16:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-11-16 11:12:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"78d566070dccd025e5499629d2dd0307f0e22654","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-11-08 16:10:51.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.\n\nThe following files contain Git conflicts:\n* manifests/controller.pp","accounts_in_message":[],"_revision_number":1},{"id":"3908e1b68b89417383a85f1f85587b625e9d1cb1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-11-08 16:12:44.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"81d163ee2d19940012482a6db45ef896ab23722f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-08 16:32:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4010bc117bea40f1adcc0001003fc6b8\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a646ded803e3460d8dc83a1173fc3c47 : FAILURE in 2m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc9d73f5244847cfb0213c7bf75e54de : SUCCESS in 3m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/912728beae2a4ecd85e294be02184ff1 : FAILURE in 8m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67b43563fa934d828d4b619cb657c6eb : FAILURE in 5m 54s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e135a7c7c10d415bb5269d14158ef888 : FAILURE in 5m 57s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/a417094ce96e4ee7886b3f3df69a8d89 : SUCCESS in 14m 17s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/933e073c3ed447b39229ad06534defed : SUCCESS in 6m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea62d524120a4e71a8b41fe90855cab9 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"d2e6a195b332b29357961d8c0d6eecd2a898b27a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-11-10 20:03:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5c9d781d29abf46f0d31a679d105824bdc20f531","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-11-10 20:06:15.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"58e2461de603cae9e525d01a438c2f6a629e3c79","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-11-10 20:10:31.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"48b5955f6365d99237d19fdb7e0ac8cfae6bafa9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-11-10 20:14:59.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"34441f332234738e8746a509eb0fad3dbbc4a2a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 20:31:13.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/385d6b808874450ca5e0137b048d9e0b\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86debe65c29f47b4a5c4f2d94bfe1dc7 : SUCCESS in 2m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cefabed610954d0caf2781a63cab32ad : SUCCESS in 3m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d2752851d85644f9b59949218dc3ec61 : SUCCESS in 7m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/992afe7b87814f57a55ead3404eb0490 : SUCCESS in 7m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/078ec8e9207d429086e84bce815b7697 : SUCCESS in 5m 24s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/a0002184ff9c4c99b70eedf83ac1af62 : SUCCESS in 14m 41s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d287a2c49b4b425e816ac42ff24765ce : SUCCESS in 7m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db723afdc17f456da0ab28c8f56f7d72 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":6},{"id":"31167f3ef04f44549a9485f2936087e3fdf06030","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-14 03:46:10.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7ffe5e9e8ba4135b506feb7452bf58ce1c7389f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 03:59:47.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ea2997be5d404986884f4f58a5ed8375\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c270aff262a4e6f9386a1918734a5a8 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea6572b978fc45bcb3ce9dd7b05dcb13 : FAILURE in 3m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c21eb75afaf43cb82455d25f5e529ce : FAILURE in 9m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebf8fcd66325490099587af91a4c06d3 : FAILURE in 6m 24s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6184d830b994a649e28ac5988ccd032 : FAILURE in 6m 28s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/69fe4780d60148f2af18fdfd38ed13c1 : FAILURE in 9m 47s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5e1826b863245d4a6871cd821fb5d38 : FAILURE in 7m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1dcc1de83c794291a7cc4f56b80ebe07 : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":7},{"id":"2bc03b4cb883ae8f35599e461b7ef593b04b01c8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-14 05:08:05.000000000","message":"Topic ovn-ofctrl-wait-before-clear-stable/yoga-stable/xena-stable/wallaby removed","accounts_in_message":[],"_revision_number":7},{"id":"a1eb8212e375e36c3a21070493f84992ac50d04e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-14 05:08:09.000000000","message":"Topic set to ovn-ofctrl-wait-before-clear-stable/wallaby","accounts_in_message":[],"_revision_number":7},{"id":"b249172bd6d084a54f4f056bf0aeee7c14b34d63","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-14 05:09:01.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"112a69cae5b9c1b2e7ccd20b44ffa1f1f50bd537","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-14 05:09:06.000000000","message":"Topic ovn-ofctrl-wait-before-clear-stable/wallaby removed","accounts_in_message":[],"_revision_number":8},{"id":"a2daa2fdd855f1107c54969b9970fd8c1f77cec7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-14 05:09:10.000000000","message":"Topic set to ovn-ofctrl-wait-before-clear-stable/wallaby","accounts_in_message":[],"_revision_number":8},{"id":"6a9fd975b4882ac99f72b27976fd472934167db2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-14 05:29:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c02e3ba141b04a95b85c33f087faeeb0\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e362452bcc64e34a0386dd35f469bb1 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6e6ff749762470fbc4bdc728fd23869 : SUCCESS in 2m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/34d314ec2bd04a918c2968b19cb88922 : SUCCESS in 10m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b00157576eb4cd885690fab5873cd60 : SUCCESS in 5m 04s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf388398f66f463c9ecaa808e41a3256 : SUCCESS in 7m 11s (non-voting)\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/49ec0a2688984eef855e9ded3692a2e0 : SUCCESS in 10m 59s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4ac95674dc14d62a656b65e91041903 : SUCCESS in 7m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f178f52fcfa40c98f4ed7f3a54804fd : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":8},{"id":"b3894bd3b8f7d63c1b463794530cff7993ab6b7f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-14 15:16:18.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"40567c78271c9717885ee599a74ed182f85df843","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-15 09:02:05.000000000","message":"Topic ovn-ofctrl-wait-before-clear-stable/wallaby removed","accounts_in_message":[],"_revision_number":8},{"id":"6e36dde6c610cfd25104cceafe00e8f7568aeb41","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-11-15 09:02:09.000000000","message":"Topic set to ovn-ofctrl-wait-before-clear","accounts_in_message":[],"_revision_number":8},{"id":"60a50186847b3316c5819aaf2db8c8487bc971dd","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"date":"2022-11-15 14:58:31.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":8},{"id":"e8011925c3aa5f0ef261924331c437559336f0b6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-16 11:12:10.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"18cc3b440e75caab33db95c92d0fce026acc237a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 11:13:11.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"012c0b057edf005697ebe97626a2fef752aeba57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 11:33:51.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3e1f9fede894a0eb2a488246bb61dae\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ad5d0254fd04f268236ae5d4822adf3 : SUCCESS in 2m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fffd641dd3db4fd49cfbd1028650884e : SUCCESS in 2m 33s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7efe17dd4f644998c18d3d931576b3f : SUCCESS in 8m 57s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aab5e45a884f41058d637f6ddea0b221 : SUCCESS in 5m 05s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/6b390ba1482744a1aebd57192d4e687b : SUCCESS in 13m 43s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89a059c61b1942e69165cf53e527b973 : SUCCESS in 6m 01s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24f253e0acff4a40a29b285d4b9649d9 : SUCCESS in 3m 02s","accounts_in_message":[],"_revision_number":8},{"id":"db3c809bd07f02c71f0f1d2662f7b68c7121981a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 11:33:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"cb5f351e7760950ade6501c61c45ee35d67ebc02","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-16 11:34:56.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47c2790c68f6405bacf0ec37130f4e22\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e07182de1b12476a9b37c93b03be996f : SUCCESS in 35s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6375bb29de745ee989a6aef86c01e2120af2d93c","revisions":{"21dda651653cfe524fcdf912981a4a0ad10d380f":{"kind":"REWORK","_number":1,"created":"2022-11-08 16:10:51.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/30/864030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/30/864030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/1"}}},"commit":{"parents":[{"commit":"df9d0e7467be0ff86ab35f253eac0311a74ab89f","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/df9d0e7467be0ff86ab35f253eac0311a74ab89f"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-01 17:45:15.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-08 16:10:51.000000000","tz":0},"subject":"Add support for ovn-ofctrl-wait-before-clear","message":"Add support for ovn-ofctrl-wait-before-clear\n\nSupport was added for this option [1] to avoid dataplane downtime\nduring ovn upgrades where schema changes have happened. This\nadds the ability for us to configure it.\n\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220808182845.2746916-2-mmichels@redhat.com/\n\nChange-Id: Ia3112b9a219b90af718c49776da49078449f09a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/21dda651653cfe524fcdf912981a4a0ad10d380f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/21dda651653cfe524fcdf912981a4a0ad10d380f"}]},"branch":"refs/heads/stable/wallaby"},"8d0ed8cb9596a3f2527f3852f6def16df0e13012":{"kind":"REWORK","_number":2,"created":"2022-11-08 16:12:44.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/30/864030/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/30/864030/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/2"}}},"commit":{"parents":[{"commit":"df9d0e7467be0ff86ab35f253eac0311a74ab89f","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/df9d0e7467be0ff86ab35f253eac0311a74ab89f"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-01 17:45:15.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-08 16:12:44.000000000","tz":0},"subject":"Add support for ovn-ofctrl-wait-before-clear","message":"Add support for ovn-ofctrl-wait-before-clear\n\nSupport was added for this option [1] to avoid dataplane downtime\nduring ovn upgrades where schema changes have happened. This\nadds the ability for us to configure it.\n\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220808182845.2746916-2-mmichels@redhat.com/\n\nChange-Id: Ia3112b9a219b90af718c49776da49078449f09a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/8d0ed8cb9596a3f2527f3852f6def16df0e13012"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/8d0ed8cb9596a3f2527f3852f6def16df0e13012"}]},"branch":"refs/heads/stable/wallaby"},"acfcb3876973edbb9d284da7901bc67556a623aa":{"kind":"REWORK","_number":3,"created":"2022-11-10 20:03:29.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/30/864030/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/30/864030/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/3"}}},"commit":{"parents":[{"commit":"df9d0e7467be0ff86ab35f253eac0311a74ab89f","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/df9d0e7467be0ff86ab35f253eac0311a74ab89f"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-01 17:45:15.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-10 20:03:29.000000000","tz":0},"subject":"Add support for ovn-ofctrl-wait-before-clear","message":"Add support for ovn-ofctrl-wait-before-clear\n\nSupport was added for this option [1] to avoid dataplane downtime\nduring ovn upgrades where schema changes have happened. This\nadds the ability for us to configure it.\n\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220808182845.2746916-2-mmichels@redhat.com/\n\nChange-Id: Ia3112b9a219b90af718c49776da49078449f09a2\n(cherry picked from commit b5d38dcbd4cc8a56cc0d028f982accf23a19b15c)\n(cherry picked from commit $eventual_yoga_commit)\n(cherry picked from commit $eventual_xena_commit)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/acfcb3876973edbb9d284da7901bc67556a623aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/acfcb3876973edbb9d284da7901bc67556a623aa"}]},"branch":"refs/heads/stable/wallaby"},"9552e3130422065a3956a5451dea6f03eb5c69c5":{"kind":"REWORK","_number":4,"created":"2022-11-10 20:06:15.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/30/864030/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/30/864030/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/4"}}},"commit":{"parents":[{"commit":"df9d0e7467be0ff86ab35f253eac0311a74ab89f","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/df9d0e7467be0ff86ab35f253eac0311a74ab89f"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-01 17:45:15.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-10 20:05:38.000000000","tz":0},"subject":"Add support for ovn-ofctrl-wait-before-clear","message":"Add support for ovn-ofctrl-wait-before-clear\n\nSupport was added for this option [1] to avoid dataplane downtime\nduring ovn upgrades where schema changes have happened. This\nadds the ability for us to configure it.\n\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220808182845.2746916-2-mmichels@redhat.com/\n\nChange-Id: Ia3112b9a219b90af718c49776da49078449f09a2\n(cherry picked from commit b5d38dcbd4cc8a56cc0d028f982accf23a19b15c)\n(cherry picked from commit $eventual_yoga_commit)\n(cherry picked from commit $eventual_xena_commit)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/9552e3130422065a3956a5451dea6f03eb5c69c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/9552e3130422065a3956a5451dea6f03eb5c69c5"}]},"branch":"refs/heads/stable/wallaby"},"92ad223363e034aa216d6971fd7136aaddcfd491":{"kind":"REWORK","_number":5,"created":"2022-11-10 20:10:31.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/30/864030/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/30/864030/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/5"}}},"commit":{"parents":[{"commit":"df9d0e7467be0ff86ab35f253eac0311a74ab89f","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/df9d0e7467be0ff86ab35f253eac0311a74ab89f"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-01 17:45:15.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-10 20:10:21.000000000","tz":0},"subject":"Add support for ovn-ofctrl-wait-before-clear","message":"Add support for ovn-ofctrl-wait-before-clear\n\nSupport was added for this option [1] to avoid dataplane downtime\nduring ovn upgrades where schema changes have happened. This\nadds the ability for us to configure it.\n\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220808182845.2746916-2-mmichels@redhat.com/\n\nChange-Id: Ia3112b9a219b90af718c49776da49078449f09a2\n(cherry picked from commit b5d38dcbd4cc8a56cc0d028f982accf23a19b15c)\n(cherry picked from commit $eventual_yoga_commit)\n(cherry picked from commit $eventual_xena_commit)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/92ad223363e034aa216d6971fd7136aaddcfd491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/92ad223363e034aa216d6971fd7136aaddcfd491"}]},"branch":"refs/heads/stable/wallaby"},"55946dcb74f5229df841f3b40d0c8db646682dc2":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2022-11-10 20:14:59.000000000","uploader":{"_account_id":5756,"name":"Terry Wilson","email":"twilson@redhat.com","username":"otherwiseguy"},"ref":"refs/changes/30/864030/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/30/864030/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/6"}}},"commit":{"parents":[{"commit":"df9d0e7467be0ff86ab35f253eac0311a74ab89f","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/df9d0e7467be0ff86ab35f253eac0311a74ab89f"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-01 17:45:15.000000000","tz":-300},"committer":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-10 20:14:59.000000000","tz":0},"subject":"Add support for ovn-ofctrl-wait-before-clear","message":"Add support for ovn-ofctrl-wait-before-clear\n\nSupport was added for this option [1] to avoid dataplane downtime\nduring ovn upgrades where schema changes have happened. This\nadds the ability for us to configure it.\n\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220808182845.2746916-2-mmichels@redhat.com/\n\nConflicts\n   manifests/controller.pp\n\nChange-Id: Ia3112b9a219b90af718c49776da49078449f09a2\n(cherry picked from commit b5d38dcbd4cc8a56cc0d028f982accf23a19b15c)\n(cherry picked from commit $eventual_yoga_commit)\n(cherry picked from commit $eventual_xena_commit)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/55946dcb74f5229df841f3b40d0c8db646682dc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/55946dcb74f5229df841f3b40d0c8db646682dc2"}]},"branch":"refs/heads/stable/wallaby","description":"Edit commit message"},"8551c877af6a2af072a31085dd7e6b56293f76a4":{"kind":"REWORK","_number":7,"created":"2022-11-14 03:46:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/864030/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/30/864030/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/7"}}},"commit":{"parents":[{"commit":"df9d0e7467be0ff86ab35f253eac0311a74ab89f","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/df9d0e7467be0ff86ab35f253eac0311a74ab89f"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-01 17:45:15.000000000","tz":-300},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-14 03:45:46.000000000","tz":540},"subject":"Add support for ovn-ofctrl-wait-before-clear","message":"Add support for ovn-ofctrl-wait-before-clear\n\nSupport was added for this option [1] to avoid dataplane downtime\nduring ovn upgrades where schema changes have happened. This\nadds the ability for us to configure it.\n\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220808182845.2746916-2-mmichels@redhat.com/\n\nConflicts:\n\tmanifests/controller.pp\n\tspec/classes/ovn_controller_spec.rb\n\n(yoga to xena)\nConflicts:\n\tmanifests/controller.pp\n\n(xena to wallaby)\nConflicts:\n\tmanifests/controller.pp\n\nChange-Id: Ia3112b9a219b90af718c49776da49078449f09a2\n(cherry picked from commit b5d38dcbd4cc8a56cc0d028f982accf23a19b15c)\n(cherry picked from commit a8f8fe2cece2e20ee8c1c1b3e7d78cb3d780d6b9)\n(cherry picked from commit b41d052d0f959b1b657c920cb7aa6166673e3963)\n(cherry picked from commit 3cb12653523459afbe9cfdddfba812e0e16788c8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/8551c877af6a2af072a31085dd7e6b56293f76a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/8551c877af6a2af072a31085dd7e6b56293f76a4"}]},"branch":"refs/heads/stable/wallaby"},"6375bb29de745ee989a6aef86c01e2120af2d93c":{"kind":"REWORK","_number":8,"created":"2022-11-14 05:09:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/864030/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/30/864030/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/30/864030/8"}}},"commit":{"parents":[{"commit":"df9d0e7467be0ff86ab35f253eac0311a74ab89f","subject":"setuptools: Disable auto discovery","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/df9d0e7467be0ff86ab35f253eac0311a74ab89f"}]}],"author":{"name":"Terry Wilson","email":"twilson@redhat.com","date":"2022-11-01 17:45:15.000000000","tz":-300},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-14 05:09:01.000000000","tz":0},"subject":"Add support for ovn-ofctrl-wait-before-clear","message":"Add support for ovn-ofctrl-wait-before-clear\n\nSupport was added for this option [1] to avoid dataplane downtime\nduring ovn upgrades where schema changes have happened. This\nadds the ability for us to configure it.\n\n[1] https://patchwork.ozlabs.org/project/ovn/patch/20220808182845.2746916-2-mmichels@redhat.com/\n\nConflicts:\n\tmanifests/controller.pp\n\tspec/classes/ovn_controller_spec.rb\n\n(yoga to xena)\nConflicts:\n\tmanifests/controller.pp\n\n(xena to wallaby)\nConflicts:\n\tmanifests/controller.pp\n\nChange-Id: Ia3112b9a219b90af718c49776da49078449f09a2\n(cherry picked from commit b5d38dcbd4cc8a56cc0d028f982accf23a19b15c)\n(cherry picked from commit a8f8fe2cece2e20ee8c1c1b3e7d78cb3d780d6b9)\n(cherry picked from commit b41d052d0f959b1b657c920cb7aa6166673e3963)\n(cherry picked from commit 3cb12653523459afbe9cfdddfba812e0e16788c8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/6375bb29de745ee989a6aef86c01e2120af2d93c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/6375bb29de745ee989a6aef86c01e2120af2d93c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
