)]}'
{"id":"openstack%2Fpuppet-ovn~865787","triplet_id":"openstack%2Fpuppet-ovn~master~Idb92e3d4f52eefabca992c027f1576700fc63234","project":"openstack/puppet-ovn","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-12-01 01:21:04.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Idb92e3d4f52eefabca992c027f1576700fc63234","subject":"Remove logic to support $::os_service_default","status":"MERGED","created":"2022-11-28 05:21:44.000000000","updated":"2022-12-01 03:41:59.000000000","submitted":"2022-12-01 03:40:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"865787","meta_rev_id":"7441d4b6a3cb35e9bb1bd4872cbd980c33a7b0aa","_number":865787,"virtual_id_number":865787,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-01 03:40:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2022-11-29 17:03:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2022-11-30 06:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2022-11-30 06:54:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-28 06:30:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-29 17:03:32.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2022-11-30 06:54:47.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"772537e1bfc047d7d2f37ce1f9d4a2a3fb2d81ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-28 05:21:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fca6e2f2fec44922103f20d9b9c071c1db4bd247","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 06:30:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd4d8c31a03940cba91e3c59c13d87a7\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06ba269856db472a914d871eda32328a : SUCCESS in 2m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5595d95f8e4b47108b03d96f6eb0d255 : SUCCESS in 2m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f16e7dfbce14482c98c9705009772e5b : SUCCESS in 2m 55s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a4fbdde7557438baaa8bf7c4df033df : SUCCESS in 7m 39s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66cd37a6f09244f9a6e48cf7cc5d73ea : SUCCESS in 5m 31s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdfe340d097c4419ad2cbb4ffb2bbce7 : SUCCESS in 5m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5f5c94644524806a2df30b1f7f58999 : SUCCESS in 4m 59s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ff4cbb88be849e3991f7256dfcfa5c1 : FAILURE in 13m 14s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6254d0a3084485aaced945e074be145 : SUCCESS in 1h 06m 36s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd6bc048d2644058b739ca3d2249767a : FAILURE in 12m 22s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/790390075ab64659aa76b21420277c47 : FAILURE in 46m 51s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c50577c768f749ddae3c8e58a8b3dc6b : SUCCESS in 7m 13s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/c149523f4a2f4821bbcac74a5e92f0f7 : SUCCESS in 11m 02s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/355eb2853d7246929d8acacb1b0a0457 : SUCCESS in 5m 49s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4681045c55bd421d9d10555b3a41adcc : SUCCESS in 5m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d0b90f0c912472db19c442d2ada7f02 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"3b78dcff2871e34ce19de86a03e1dba8cad05ac3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-28 07:54:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7348675840fb5f75b99dad6d0ac69f48add011d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-28 09:05:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebcc689ba3b04d58b3c62ec990992dd2\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27c6825ae3a4482aae1c5d5529224e54 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19bdf488c34240748e268079b807ca07 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4adb1e8202184115aa868653417709f6 : SUCCESS in 3m 18s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b5c5976262448afaff9954117e438f2 : SUCCESS in 10m 09s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6ed7a11197148c3a88cae675bddd440 : SUCCESS in 7m 03s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b7ca788b04a4f6ca94ea0b1d94a78d1 : SUCCESS in 7m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/770d49e8e3d140598afe8edca2ad1071 : SUCCESS in 7m 06s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f519f822b8084f5a944036f38ac6859a : FAILURE in 15m 20s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a880a0122fd497e9d393a7518cf71ec : SUCCESS in 1h 07m 37s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f1c6cb3c7694919b8b5c359ea769237 : FAILURE in 15m 24s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/70d3cbb11c944e2ea513b9efd4b1713d : SUCCESS in 53m 14s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea5c3ff8bf594c2aa9c701edaf3fdf85 : SUCCESS in 7m 57s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/fea8a32758bf4c2eade1ab26fe16eb70 : SUCCESS in 13m 29s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b640da61b604407984292bce8b0ed5d : SUCCESS in 8m 06s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c53a338c445e4af68c1c914f47c38622 : SUCCESS in 6m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fddbeb4adc7c46f1b487b6749cb32832 : SUCCESS in 2m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d79a45e8414365b12133c6f821ef3dde6723dbe2","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-11-29 17:03:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"45a03c54aa31258ea5adde2201fa0ccbc13a95dd","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-11-30 06:54:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e40b97a3b8a4ab3187dcf3a5b8535a6a62247586","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 06:55:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32b394e5923d3e75fb90986e1156616b7127da4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 08:04:52.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5d2977fd5e8943ff8b68d9b9dd7b7b83\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f6c2b8715ff482a9a0354c9a2189178 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a044a5be6124fea87ed3543483e9f73 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58b18fadc8e74f469f6064bc5a0616cc : SUCCESS in 3m 56s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a03fe994fe9408f91b49f7089d9384b : SUCCESS in 10m 45s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/613d6c406d6144abb9d8c907fa76d19d : SUCCESS in 7m 29s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8db3b0b4997a46f99314584ad92a6145 : SUCCESS in 8m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/25994bb9b81d47f98a049969701f573f : FAILURE in 59m 24s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/169aed0ae99c45d2b498289bf439df39 : SUCCESS in 48m 38s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/0630374ddaa64cc2bd6668c38226c637 : SUCCESS in 15m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce95ab90ce7b4e53a236efc1ce2582d2 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"41ecba66d30d8d49fad9523103ce9a7f899dba9a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-11-30 09:23:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a25009f48f3b769bbecb342848c8919bd778e168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-30 10:41:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/93048f97d5f04caba65c1ea197af7b59\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf09d6ca0eab4361a047419e6a0ef95d : SUCCESS in 2m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/355a05d1e4c44545acb2a200331b9bc8 : SUCCESS in 2m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6b9c7a3c5754474b78a4de73a50e079 : SUCCESS in 2m 46s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5580fc12a4c420080169beaa17cb003 : SUCCESS in 9m 23s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/027923aa40674dc7b164988615a89c27 : SUCCESS in 6m 03s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/80ca33f2579742198cf71ac9792c0d40 : SUCCESS in 5m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/453e859d0aa3475fba3efcfc2ee64093 : SUCCESS in 6m 10s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1c9adad1c1145b9be8c303d9ea017c0 : FAILURE in 16m 56s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/05f4775cd6bb4af39b886b8c98e6b7a7 : SUCCESS in 1h 12m 27s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/593cc87d7b754c7f92713de469d9f8f5 : FAILURE in 14m 28s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/282756c9c85040ff8f5c8c41bdcf3bdb : FAILURE in 53m 43s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52c7a9a9c93a43d2baa4e5fe6bbdde1e : SUCCESS in 6m 29s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/746b097e3f9d4631b97d757c0568fae8 : SUCCESS in 12m 42s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0dace8283d444aa79caf0c5169f599a3 : SUCCESS in 8m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10aaf4b6b8564984980d5c962f079032 : SUCCESS in 6m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be60a150998f440a8195e1057d8d4e2d : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"51cfd2c080cbc65fee6f8d087c2150a2309f81d5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-12-01 01:21:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"42155b8a148b0876c0af828266a30cc154644673","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 02:28:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/231b82f593194fa6b794ad41469fb5a9\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81a77644aecc424abb8d79cc8cf700ae : SUCCESS in 2m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72fa0d21f3094decb2b1bc5b6823178b : SUCCESS in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ef5bd2b92eb4ce7b9ad9a06c1c4b2ad : SUCCESS in 3m 00s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/e793ef6b6f7842a1acc0598330cd7a9e : SUCCESS in 10m 26s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d7e34af646247f7b84bfbadf68c86a3 : SUCCESS in 6m 52s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de0b727a546d448482a946577b7a1af8 : SUCCESS in 5m 08s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a83cc3a85a042b4af4bafc8f0f654f9 : SUCCESS in 5m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6388e32361f4cb5972c6bbe2582474c : FAILURE in 12m 51s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1ca86b482b14057824bf978c26694fb : SUCCESS in 1h 05m 20s\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bb6c1474a54497e839b10694aea336f : FAILURE in 10m 44s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2968400208df4eaa8d1129768b9ffb4c : SUCCESS in 47m 22s\n- puppet-openstack-litmus-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa425910df1e49038d6057e3780fbb54 : SUCCESS in 6m 24s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/4dbe7e64509e473ea38f9dea90138df4 : SUCCESS in 12m 21s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/191770d527eb4e6fafb8cdb9cae235d4 : SUCCESS in 5m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bdb94efb25f46258547a7f0b1737eec : SUCCESS in 5m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f09b614af57549d18673d5bd4d6b9308 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":1},{"id":"9b8a3221d72fc46df4f9d5ae70d0f9521d0b7506","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 02:30:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e371a0ddf1de583471090778cd9e41f55ffaf2e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 03:40:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a51fb01768b455d8e26652691672790\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbcc06d8776c4ce5b20be066ee612e99 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfb8a88f6420436eb183a83421df37de : SUCCESS in 4m 08s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04150e43288f4ab6aac317486d8865a5 : SUCCESS in 3m 07s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/91bf418ae0a5430ba2600c204cfa7430 : SUCCESS in 8m 29s\n- puppet-openstack-unit-6.28-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50a46b86836a4fb99899c844f6789027 : SUCCESS in 6m 34s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce8430e9294e4b5f8b3dd6e56ab01b51 : SUCCESS in 4m 57s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/71a38c89fc4544c89d611783be67276b : SUCCESS in 1h 05m 33s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/185209c51445458cbd6597f498006fbe : SUCCESS in 59m 01s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/1a7988debb2c4f548f837d55a7be3a8a : SUCCESS in 14m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51026f3b43554d96b2d1117007923f01 : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"7d321357cdfb73095cbbd813456f69ebacca2196","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 03:40:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7441d4b6a3cb35e9bb1bd4872cbd980c33a7b0aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-01 03:41:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1dc1f9f7ff64792a42a8737caacbee2\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9666d96f8ba2431181d536ac960d4a99 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb9fed3f7aca11dc1b4d6a3867ae4d367b53b671","revisions":{"cb9fed3f7aca11dc1b4d6a3867ae4d367b53b671":{"kind":"REWORK","_number":1,"created":"2022-11-28 05:21:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/87/865787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/87/865787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/87/865787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/87/865787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/87/865787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/87/865787/1"}}},"commit":{"parents":[{"commit":"b5d38dcbd4cc8a56cc0d028f982accf23a19b15c","subject":"Add support for ovn-ofctrl-wait-before-clear","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/b5d38dcbd4cc8a56cc0d028f982accf23a19b15c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-28 05:20:37.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-11-28 05:21:36.000000000","tz":540},"subject":"Remove logic to support $::os_service_default","message":"Remove logic to support $::os_service_default\n\n... because support for $::os_service_default was deprecated[1] during\nthe previous cycle.\n\n[1] 5c809de01630ee7eede500106143810ca57b58b5\n\nChange-Id: Idb92e3d4f52eefabca992c027f1576700fc63234\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/cb9fed3f7aca11dc1b4d6a3867ae4d367b53b671"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/cb9fed3f7aca11dc1b4d6a3867ae4d367b53b671"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[]}
