)]}'
{"id":"openstack%2Fpuppet-ovn~898273","triplet_id":"openstack%2Fpuppet-ovn~master~I8d01236c04cf2b61f1c1bf39ecf54ca0a67a33a9","project":"openstack/puppet-ovn","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-10-31 01:12:23.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"I8d01236c04cf2b61f1c1bf39ecf54ca0a67a33a9","subject":"Replace params by hieradata","status":"MERGED","created":"2023-10-15 08:25:24.000000000","updated":"2023-10-31 03:55:38.000000000","submitted":"2023-10-31 03:55:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898273","meta_rev_id":"630db0f2c99c2f24fe6cc9452fc33b26872d5e8a","_number":898273,"virtual_id_number":898273,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-31 03:55:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2023-10-26 16:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2023-10-30 08:08:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2023-10-30 08:08:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-15 08:55:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-26 16:10:27.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2023-10-30 08:08:55.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"97d84f8e09193e771209e218c66b86fc3463a8d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-15 08:25:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"889e157c2d2ea8984b13ec7bf78f09d6dfcb5f26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-15 08:28:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"01cc1d1e78b64e9f99999b5fa3c19d5d759867ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-15 08:55:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e984e89dd0f48d081d7638152627bad\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e928548168094046bffbf8c3569fdac2 : FAILURE in 3m 18s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6225ab7c79004971a94545d3cef0d939 : SUCCESS in 3m 12s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/67faeb27899945c78ec9d4fd5db9c89d : SUCCESS in 5m 19s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5d062d9826b46f4a90db021923ce574 : FAILURE in 8m 50s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4e0074b58ee4c5cb1300fbe24c92557 : FAILURE in 4m 51s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/7663ed8032144a34b33cc7f7a831671e : FAILURE in 8m 24s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e1ca7c9ed7994badab0ce308a464843d : FAILURE in 5m 08s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5350b827c3144486a7b4c31341fb9745 : FAILURE in 6m 48s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/60b345cc5d62447692a5c8f089be9a67 : FAILURE in 11m 47s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e197cc93eae44d4585e45c5ea5cd6c0e : FAILURE in 6m 18s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/61a61139ce3746b7988c8ab77d5a858e : FAILURE in 8m 21s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fe54f4ec2b64602a7d367f54f2c4b61 : FAILURE in 7m 48s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b7915721563485ea17d8be5e3fdc552 : FAILURE in 5m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea6c8066d23f454c983204aabeb7e2e4 : FAILURE in 6m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f180fbcc718b3cd1ebecdff641dc68ca939dcb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-18 13:49:13.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9fd6a9a1d0d9b90086eed779115263b301a6ec55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-18 13:49:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"307818eea9b5f30ba257101b35b3dcc4eb137350","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-18 14:06:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c48109d0553421d8e69f533ab1d6b1c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af0d6bb253104e548d7381acb758b0a1 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00867234d8434268a9458c9bd49e40ad : SUCCESS in 2m 44s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ee233dbad4c412ab051d5e08220cfb4 : SUCCESS in 6m 09s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/690b18d169764ac49aff24e3baec980e : FAILURE in 9m 36s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10e83b24095d47aa95b4bb9b53aa04f6 : FAILURE in 5m 35s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/9a5041f50aa24e1d92e0e9dfba672528 : FAILURE in 8m 51s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68344d11bd3d450fafbab44664265b2a : FAILURE in 5m 46s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34774355e64a41958cd78652fe28040b : FAILURE in 6m 05s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/220b4758d8294e6c88d92fca5a2dfdaa : FAILURE in 9m 38s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c5e1496efe4404d8684b0f9675519da : FAILURE in 9m 21s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c96da9a9f6c34e45824f7faaeebf557d : FAILURE in 10m 25s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a78f41c11eb4e0093fe6d6e2e8be95b : FAILURE in 9m 08s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08ece609e0bb4c55bee271f574979c96 : FAILURE in 5m 30s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff1cabbd69284e05916e3493bb73f332 : FAILURE in 6m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fce3696dac52f4fb4881c2b803bdc09ed249528d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-23 12:48:54.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c0e4a6756b94ccfed3b9e7f6226c445d2c07b5fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-23 12:50:30.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f14f2b12bec7ef2e5b7d93c5cd47306636fb1486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 13:09:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b83990f22de14ac0ad353f8f94480dcf\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f740d06438a3453bb5009541399498bc : SUCCESS in 2m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2559f82e948c4b99ab14a12a55de74c5 : SUCCESS in 2m 44s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/b69bc96e71534a0a9f1e65defb93e13e : SUCCESS in 5m 00s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/70029be02ce4435a9c5009ec4103820b : FAILURE in 8m 59s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52aae0c5472e4366ab476cdeb0977aa6 : FAILURE in 5m 52s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/09e80191af16420cb372f606c228ffc8 : FAILURE in 9m 43s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/653254aff7cd4bf39c5aca76ed849b12 : FAILURE in 5m 59s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8124f7fa432f4d469fafc0519c493aed : FAILURE in 6m 42s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/737742a6fe104bc3906c8307c62de311 : FAILURE in 11m 57s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e9a88e790eb423488a2146ae65a2e53 : FAILURE in 7m 43s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e5cd4f81ba94084bd6a2e3bd6a9e8ab : FAILURE in 12m 44s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/357fb2585e794d0c9b3cb86eb051c779 : FAILURE in 11m 18s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e28751db59446568ce33515b2356df6 : FAILURE in 8m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f13ef006dfeb4b8abf44d6d67f988206 : FAILURE in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1fa4dfb529bd7361fc68e49596e246c7280fb8d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-23 13:54:01.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"559d047e04293292aeb68f7b3aff01ab5192c033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-23 14:30:55.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"8a375595dafafbcdb9d7c55f58d6ab2a89edd664","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-23 15:48:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d88ac32a08a4f2ead58712ecb5dab90\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e9151fb34784196992977bdc0616bb2 : SUCCESS in 2m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/593a7454ec554fa9b2bb93f5c8f40340 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/ead78a5721bd45cdbe0860893155b4ac : SUCCESS in 5m 08s (non-voting)\n- puppet-openstack-unit-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/847ac52879d64ec882902aa8af5d1ef0 : SUCCESS in 7m 47s\n- puppet-openstack-unit-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18063e1bfc244d589dd521f98aa645b2 : SUCCESS in 5m 34s\n- puppet-openstack-unit-8.0-centos-9-stream https://zuul.opendev.org/t/openstack/build/e8498c328cc64b68b1d2ea96c35c71fa : SUCCESS in 7m 42s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a3e011e19734a1881bc4312d9ee1393 : SUCCESS in 5m 32s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30c33e4d3090446c9c087df159c3ac95 : SUCCESS in 36m 23s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/632991c33bd945b6bcf09fb781f4b602 : SUCCESS in 1h 12m 12s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae011b637f84411c90e840e88d8987ef : SUCCESS in 33m 27s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5039fbb858d347d99183887f0e179093 : SUCCESS in 54m 45s\n- puppet-openstack-litmus-7.22-centos-9-stream https://zuul.opendev.org/t/openstack/build/cf93330b56d14d7aa9b09db0dc6562b4 : SUCCESS in 10m 05s\n- puppet-openstack-litmus-7.22-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cca2406ba54649ffb690608af74a12c9 : SUCCESS in 6m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a004c0c5a9d416db53eb99d5f5f3624 : SUCCESS in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"231fae92ff4c97e6f77238e6e08d1cfc71c5a392","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2023-10-26 16:10:27.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"875753693a179885697a68a60e0af17198860603","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-10-30 08:08:55.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"473ede2d2768285580960a1b4f2c12b707662454","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 08:09:07.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5628ca5766554c9fb8aacf328ec1bdbf\n\n\nWarning:\n  Change 886112 in project openstack/puppet-vswitch does not share a change queue with 898273 in project openstack/puppet-ovn","accounts_in_message":[],"_revision_number":8},{"id":"cf17920d8f2c333ec3150c8dd2a8a7e5dc759d2b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-10-30 08:11:06.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d1cb3962a49e3d15f017b1f43e2ea86840fd9adf","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2023-10-30 08:11:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"eecdaa1254ab365161941c452eb5d76a939c5bab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 09:17:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/637b34441dcc472ca49e4eaf0206e890\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2d996563f2de4296a1db500fb37086f0 : SUCCESS in 2m 27s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80dfdb55ab294855a913362b1ef5ff9c : SUCCESS in 2m 44s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ccfb44640de4f7199c590fc1213b497 : SUCCESS in 5m 10s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7c77a008dea4852b6cfe6d536a3d99c : SUCCESS in 7m 49s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0bdf3233cfea4a4dbb466266dfb55f9d : SUCCESS in 5m 52s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/48996921ae2d4d70947e01649743527a : SUCCESS in 10m 48s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02e2401e20e441a09fc4ef558057f23d : SUCCESS in 5m 10s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7619c54915a341538e25eab43c4a9f6f : SUCCESS in 39m 43s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/208445dc02474ef795bb46a6e63c42cb : SUCCESS in 1h 02m 54s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81c5796c0409448f99d847e273c9d0cc : SUCCESS in 37m 14s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/726e7267e0cb484a98051d15ead89512 : SUCCESS in 47m 35s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0730b9d0f694e84abef2764920fe43e : SUCCESS in 11m 01s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b49f19befe64dccbf99dc7a128436ce : SUCCESS in 6m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d644fbca8d0545dd9f1602e8e400e1ac : SUCCESS in 7m 41s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"738c89323bb69a694273da44ba422e9c1830a317","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 09:17:23.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1fadd9b5966439e9d7dd1a9f4e0cd48\n\n\nWarning:\n  Change 886112 in project openstack/puppet-vswitch does not share a change queue with 898273 in project openstack/puppet-ovn","accounts_in_message":[],"_revision_number":8},{"id":"2cd77fb23dd3cea2e9b92c77a176ae433de46b33","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-30 14:03:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"63fb1b57dfcf4e9e0d0061528515d949ad8646e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 16:37:53.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6e5f06df4c54c51859fc9e120f843be\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/091d5f6e7a834654ae113f70fe8af9e1 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e27883b03a96452186ed1387d6d6e7f1 : SUCCESS in 2m 33s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/446d5bee733c42f2a85b9a43d11e6aef : SUCCESS in 5m 29s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a23f161dc064b9785d87cb633266cea : SUCCESS in 10m 41s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/039cca7fdba54f0a8ccee5cb0a9bc0df : SUCCESS in 5m 36s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/528db0ffb5244184820242008afe328d : SUCCESS in 8m 53s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/130a305de4e740fb993643f0f311481d : SUCCESS in 6m 36s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9d208c0eade4422a39fac9b97814ac1 : SUCCESS in 42m 21s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e570a6dbc2aa4f81a0512476fe70ad58 : SUCCESS in 1h 05m 25s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b8ba395c9a045bd80c429ca2876b20f : SUCCESS in 38m 01s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7aa4862a84514336a6a38b7940c33aa1 : SUCCESS in 53m 24s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/e616b14c3cf141cea0e9e2367c863f20 : SUCCESS in 12m 33s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7e27a788f1a4226a0a774828ab6871b : SUCCESS in 6m 18s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4714eef4a4214b589e7af8ae21c4e15c : SUCCESS in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2a6ba2749e4c515e68dd2dc00dc41aad4e58357e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 16:38:44.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"72d91003f2e5ea43be8cbb9406fbb286a1379859","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 17:48:39.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed9e25af81ab4703b1d055dba8192ea7\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81a4e1ffd05444ce84880e1733eb7d57 : SUCCESS in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09a6e7c38b6648919cc6f3f8f8a722ae : SUCCESS in 2m 44s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/432ea1a8bb924ffa8d4a537dccc8250d : SUCCESS in 11m 11s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41472b72e33e45d29a0a05850273225e : SUCCESS in 6m 00s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c1424440ab54d27a5196f7eaf423131 : SUCCESS in 11m 35s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/32c7ec5b582645f291d5275cec11de85 : SUCCESS in 1h 01m 39s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d5a096a04e7407caf2f015e11b9acae : FAILURE in 1h 06m 56s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/03e1e75032884660afdbb9832cca9821 : SUCCESS in 11m 35s","accounts_in_message":[],"_revision_number":8},{"id":"c786baaecb3f9254a70ebf300c01e53ae027f603","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-31 01:12:23.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bd2bf7897dd5e39f9fc63514f7e9480b21d45963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 02:09:56.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44ec4e5d667144c899a92f8d14a96fd2\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad54f4b63d9d4dfc8da14d3ddc64f2ec : SUCCESS in 2m 22s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a78c3958bf7344fa977a309ae5979614 : SUCCESS in 2m 25s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/67d0c066ec804bb8b6a772dcadba7d85 : SUCCESS in 6m 06s (non-voting)\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/3056000d69e54f5a906cb8a21741c15d : SUCCESS in 7m 31s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a1b0f95c3d541d2835f31c98cbb902b : SUCCESS in 5m 21s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/576be4b772034caeb40eac0e7e469691 : SUCCESS in 8m 18s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87eb2a9b44e242cc89a235638ef24b87 : SUCCESS in 5m 24s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e417eb42f7ab4516b4a2a6ad6720da71 : SUCCESS in 41m 04s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d116e8301d6b40ecb1420c05da35bfb8 : SUCCESS in 51m 41s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/163f02c40d7c4245ab893c11aec74d48 : SUCCESS in 36m 09s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/46cf5ff99f284cf58c27cbbbd8438243 : SUCCESS in 41m 02s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9401468d50a4599a52ea902ffcf20c4 : SUCCESS in 10m 59s\n- puppet-openstack-litmus-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6327c8edf5e042e58a414f0a02350098 : SUCCESS in 6m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4514ac455dd34af0b67678d40fc96331 : SUCCESS in 6m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8db5a91808fee92c329f7639c0a90537a18ccbb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 02:16:23.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"47822735f5a73dcd2c967f0560babb3602c44577","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 03:55:38.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/950070853df14a0bb99f38abdf1cff8b\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73e684dffdcf429bb21f78a0a1402919 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7f761f500ed48b68bc7403c1b50c463 : SUCCESS in 3m 04s\n- puppet-openstack-unit-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd81cbfd5c204edd8f515cea4ea79b33 : SUCCESS in 10m 45s\n- puppet-openstack-unit-7.26-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1bef85e3564c418ebdc6e44609330a54 : SUCCESS in 7m 08s\n- puppet-openstack-unit-8.2-centos-9-stream https://zuul.opendev.org/t/openstack/build/5970be26d3334b77a49c8dd105a25bb2 : SUCCESS in 8m 52s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1373d0799bef4828b251553b484a4ec3 : SUCCESS in 1h 10m 16s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/461cffab1d9843ccbc496ee70c52758e : SUCCESS in 52m 23s\n- puppet-openstack-litmus-7.26-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d981c9626c44461b318c8c6b6e7a743 : SUCCESS in 15m 00s","accounts_in_message":[],"_revision_number":8},{"id":"630db0f2c99c2f24fe6cc9452fc33b26872d5e8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 03:55:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"02512cccbcd3d8b236bf332de97db727a469e862","revisions":{"3301dc841bf3c9e21e0e51ca7db86497ba9097f2":{"kind":"REWORK","_number":1,"created":"2023-10-15 08:25:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/898273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/73/898273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/1"}}},"commit":{"parents":[{"commit":"e6d918f61655b3a3713b4133bf6bb8761cb95168","subject":"Debian/Ubuntu: Drop old workaround for broken service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e6d918f61655b3a3713b4133bf6bb8761cb95168"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-15 08:12:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-15 08:25:11.000000000","tz":540},"subject":"WIP: Replace params class by hieradata","message":"WIP: Replace params class by hieradata\n\nChange-Id: I8d01236c04cf2b61f1c1bf39ecf54ca0a67a33a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/3301dc841bf3c9e21e0e51ca7db86497ba9097f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/3301dc841bf3c9e21e0e51ca7db86497ba9097f2"}]},"branch":"refs/heads/master"},"5a081a1ec5086d86d13070dfff97bce254fba0c2":{"kind":"REWORK","_number":2,"created":"2023-10-15 08:28:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/898273/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/73/898273/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/2"}}},"commit":{"parents":[{"commit":"e6d918f61655b3a3713b4133bf6bb8761cb95168","subject":"Debian/Ubuntu: Drop old workaround for broken service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e6d918f61655b3a3713b4133bf6bb8761cb95168"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-15 08:12:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-15 08:28:11.000000000","tz":540},"subject":"WIP: Replace params class by hieradata","message":"WIP: Replace params class by hieradata\n\nChange-Id: I8d01236c04cf2b61f1c1bf39ecf54ca0a67a33a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/5a081a1ec5086d86d13070dfff97bce254fba0c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/5a081a1ec5086d86d13070dfff97bce254fba0c2"}]},"branch":"refs/heads/master"},"b3023e6afa5c46d22cf95820f5afd674f89f6f0c":{"kind":"REWORK","_number":3,"created":"2023-10-18 13:49:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/898273/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/73/898273/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/3"}}},"commit":{"parents":[{"commit":"e6d918f61655b3a3713b4133bf6bb8761cb95168","subject":"Debian/Ubuntu: Drop old workaround for broken service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e6d918f61655b3a3713b4133bf6bb8761cb95168"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-15 08:12:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-18 13:49:05.000000000","tz":540},"subject":"WIP: Replace params class by hieradata","message":"WIP: Replace params class by hieradata\n\nChange-Id: I8d01236c04cf2b61f1c1bf39ecf54ca0a67a33a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/b3023e6afa5c46d22cf95820f5afd674f89f6f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/b3023e6afa5c46d22cf95820f5afd674f89f6f0c"}]},"branch":"refs/heads/master"},"326f731d9f50667187ec20e6784d58d30835fbd2":{"kind":"REWORK","_number":4,"created":"2023-10-18 13:49:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/898273/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/73/898273/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/4"}}},"commit":{"parents":[{"commit":"e6d918f61655b3a3713b4133bf6bb8761cb95168","subject":"Debian/Ubuntu: Drop old workaround for broken service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e6d918f61655b3a3713b4133bf6bb8761cb95168"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-15 08:12:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-18 13:49:35.000000000","tz":540},"subject":"WIP: Replace params class by hieradata","message":"WIP: Replace params class by hieradata\n\nChange-Id: I8d01236c04cf2b61f1c1bf39ecf54ca0a67a33a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/326f731d9f50667187ec20e6784d58d30835fbd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/326f731d9f50667187ec20e6784d58d30835fbd2"}]},"branch":"refs/heads/master"},"c0d357870fc028d344a1118d449a1d4b6bbcfc8a":{"kind":"REWORK","_number":5,"created":"2023-10-23 12:48:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/898273/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/73/898273/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/5"}}},"commit":{"parents":[{"commit":"e6d918f61655b3a3713b4133bf6bb8761cb95168","subject":"Debian/Ubuntu: Drop old workaround for broken service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e6d918f61655b3a3713b4133bf6bb8761cb95168"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-15 08:12:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-23 12:48:44.000000000","tz":540},"subject":"WIP: Replace params class by hieradata","message":"WIP: Replace params class by hieradata\n\nChange-Id: I8d01236c04cf2b61f1c1bf39ecf54ca0a67a33a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/c0d357870fc028d344a1118d449a1d4b6bbcfc8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/c0d357870fc028d344a1118d449a1d4b6bbcfc8a"}]},"branch":"refs/heads/master"},"d23890880fee246256e79a9ab5be16627e026047":{"kind":"REWORK","_number":6,"created":"2023-10-23 12:50:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/898273/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/73/898273/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/6"}}},"commit":{"parents":[{"commit":"e6d918f61655b3a3713b4133bf6bb8761cb95168","subject":"Debian/Ubuntu: Drop old workaround for broken service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e6d918f61655b3a3713b4133bf6bb8761cb95168"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-15 08:12:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-23 12:50:23.000000000","tz":540},"subject":"WIP: Replace params class by hieradata","message":"WIP: Replace params class by hieradata\n\nChange-Id: I8d01236c04cf2b61f1c1bf39ecf54ca0a67a33a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/d23890880fee246256e79a9ab5be16627e026047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/d23890880fee246256e79a9ab5be16627e026047"}]},"branch":"refs/heads/master"},"e0f5f05f8594405ae265d85ab303e61cfa88d785":{"kind":"REWORK","_number":7,"created":"2023-10-23 13:54:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/898273/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/73/898273/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/7"}}},"commit":{"parents":[{"commit":"e6d918f61655b3a3713b4133bf6bb8761cb95168","subject":"Debian/Ubuntu: Drop old workaround for broken service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e6d918f61655b3a3713b4133bf6bb8761cb95168"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-15 08:12:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-23 13:53:47.000000000","tz":540},"subject":"Replace params by hieradata","message":"Replace params by hieradata\n\nThe params class is the legacy approach to define OS/version specific\nvalues. This replaces the params class by the module hieradata\nfollowing the recent standard.\n\nChange-Id: I8d01236c04cf2b61f1c1bf39ecf54ca0a67a33a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e0f5f05f8594405ae265d85ab303e61cfa88d785"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e0f5f05f8594405ae265d85ab303e61cfa88d785"}]},"branch":"refs/heads/master"},"02512cccbcd3d8b236bf332de97db727a469e862":{"kind":"REWORK","_number":8,"created":"2023-10-23 14:30:55.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/898273/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/73/898273/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/73/898273/8"}}},"commit":{"parents":[{"commit":"e6d918f61655b3a3713b4133bf6bb8761cb95168","subject":"Debian/Ubuntu: Drop old workaround for broken service status","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e6d918f61655b3a3713b4133bf6bb8761cb95168"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-15 08:12:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-23 14:30:45.000000000","tz":540},"subject":"Replace params by hieradata","message":"Replace params by hieradata\n\nThe params class is the legacy approach to define OS/version specific\nvalues. This replaces the params class by the module hieradata\nfollowing the recent standard.\n\nDepends-on: https://review.opendev.org/c/openstack/puppet-vswitch/+/886112\nChange-Id: I8d01236c04cf2b61f1c1bf39ecf54ca0a67a33a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/02512cccbcd3d8b236bf332de97db727a469e862"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/02512cccbcd3d8b236bf332de97db727a469e862"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
