)]}'
{"id":"openstack%2Fpuppet-ovn~933340","triplet_id":"openstack%2Fpuppet-ovn~master~I88512683c88c403acca67d5d7c839d6fb7354557","project":"openstack/puppet-ovn","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-25 14:44:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I88512683c88c403acca67d5d7c839d6fb7354557","subject":"Support customizing inactivity probe for DB connections","status":"MERGED","created":"2024-10-24 17:17:21.000000000","updated":"2024-10-25 14:45:01.000000000","submitted":"2024-10-25 14:44:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933340","meta_rev_id":"fc0cb9c6dea3a7255082620882901b9e9185cd8c","_number":933340,"virtual_id_number":933340,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-25 14:43:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-10-25 12:20:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2024-10-25 13:53:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2024-10-25 13:53:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 18:33:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-25 12:20:41.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2024-10-25 13:53:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"c06e4d3f21be83096d4f82e0c76c6fbb1e75e1c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 17:17:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3750fe3fa9e8344ebed7c6c55e1075422b404003","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 18:33:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0e9a8d7a4c54abdab9bea6a7c29b442\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc9efc8c8ee64d09b2d3dcf2398d50d5 : SUCCESS in 2m 39s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6af959f3e394073a8c1db8f9c4b4d43 : SUCCESS in 2m 02s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/01ed4cd8ba77421b991ac1f9b671974c : SUCCESS in 3m 22s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/213bbc12bcd947fb94b03746a44caf47 : SUCCESS in 6m 39s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9334bb27c59c4e55b7a202b38399df16 : SUCCESS in 4m 52s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/e707bd8503824eafab37b3a10da3098d : SUCCESS in 7m 23s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/502e7759facb45328346648d69f90d04 : SUCCESS in 3m 43s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e57ab73b45ee40f38c63ab713bc373f1 : SUCCESS in 45m 40s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4acdc8a7450942dd94b18c94d9b593aa : SUCCESS in 56m 33s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70c88e2029a84f1ba63559e68ac859ce : SUCCESS in 42m 16s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d6ce9e27cb0408e83149308be855c18 : SUCCESS in 52m 05s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c1ebd2b32ad40868ddc433e9ab5b04c : SUCCESS in 11m 03s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c90d58463fec4a22ab83e2d33cbd5a0e : SUCCESS in 5m 06s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/089763a437114f03adbb56d195d29ce4 : SUCCESS in 4m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07a3056256d34956993c53d907ed50c8 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3cf61126c72ad18d6aebf7db2c751dd04f60f9fb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 19:33:42.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"2228b0903ae5075b2a1d78147a18ff5f26313788","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-25 02:18:24.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a4de1268cba9ba51e34fb481046a9dae6f6325e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 02:55:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79f875b7fcff42caa1ae3e6991b13c14\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83d3c1ec2e6744ceab0abc88956f7ab8 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a643fd2a91d74817b29f1c3039d3acfa : FAILURE in 2m 16s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/495480936a754421b82d16173cda480c : FAILURE in 5m 42s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e1d17749f1c4c81bdfc2b4824748eb7 : FAILURE in 7m 23s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dacaf0ea00b1419090881f9b5eb6a4be : FAILURE in 4m 27s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c63cc71dc1f4fdf8eaeef1a30a74f8e : FAILURE in 6m 54s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09bf071bdcb74f26902853dbf60a1497 : FAILURE in 4m 28s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/700803ccdb644b04a40f0ca4f60d310b : FAILURE in 6m 26s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9bc0d479f3349f5ab89caa4ab549683 : FAILURE in 11m 46s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b30b229950834d8caff439e86371f65f : FAILURE in 4m 13s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6cb6f8e814d42499bde7ac8c39d34c3 : FAILURE in 9m 54s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/464f947af60c4c87a6943b238fef8067 : FAILURE in 8m 07s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae5c41a56b3c4dd4b219f7e4cba68827 : FAILURE in 3m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b473995eccba4f14b48f2c381a553c71 : FAILURE in 3m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59fa75e809024c57b2d7b6631be17455 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":2},{"id":"1e27392235a7e30f79776a8af5d4a63aaf1e9aac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-25 03:05:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"283469c745716d1d12bce146e2d1b00583ce3400","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 04:15:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f07b8e7e1c9411b8ab9968e9f9df90c\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcb0264497ac40da9e621eec6b353c85 : SUCCESS in 2m 37s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c82a2c45dff4321a09bba294ee3d4c5 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/839ab847f0fa4238919e9662e5012dbd : SUCCESS in 5m 07s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/97d57ec6dad14aa08883bea69852bbdd : SUCCESS in 7m 21s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60b1673e37b84d5f826d33cd20ff34f9 : SUCCESS in 3m 24s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/fedcb1f9351b4a39973c0f6fbc630ef7 : SUCCESS in 7m 10s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76ffe076b97b4334b389e4010ee462bc : SUCCESS in 3m 47s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fdbc89582a2f443c9a8a89954d486546 : FAILURE in 1h 07m 40s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e04162109c3480a98240ae7157b7b6b : SUCCESS in 55m 03s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/959f3fb9a1654f849f4987364a17215c : SUCCESS in 36m 30s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ccd5447322743d5b8a7e4e3caba600a : SUCCESS in 45m 00s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/95283b68da2a4209a6f72eb3c81dbad3 : SUCCESS in 10m 08s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0184742f467f45c28979ea29ecf0cb54 : SUCCESS in 4m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8f361cbb4a94fb381644b30ce5d8dbb : SUCCESS in 4m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d621845efbe94bb48adbf2b5ff195538 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":3},{"id":"46af6cc408abc037fbfbf2db24fd38ee0ebbfeb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-25 08:57:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"07a84883c400bbe9b68ce2250b566340dd189086","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 10:05:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ed0b7dcd8ae472ea1a4e2f51901b087\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc3dcdd11e8345e2b2e3adf1d99e0cb6 : SUCCESS in 3m 26s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6753cca1c52941c78c2b1a74e125f313 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-8-centos-9-stream https://zuul.opendev.org/t/openstack/build/529378ec2b474f4082bd40482f056537 : SUCCESS in 4m 51s (non-voting)\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/b94b7f99e0db4f2fb09bd43ec2eb1343 : SUCCESS in 8m 37s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2de557d6409c4e4197881950130163ca : SUCCESS in 3m 27s\n- puppet-openstack-unit-8.8-centos-9-stream https://zuul.opendev.org/t/openstack/build/509959d30e0847878c7b064023f476b5 : SUCCESS in 5m 19s (non-voting)\n- puppet-openstack-unit-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/318721c2fdaf4a1782ce1f92c7b035c7 : SUCCESS in 4m 07s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12103e71cced484287a0b90a200ccf3e : SUCCESS in 43m 01s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/986f02c2624645d5b58a60b0ac0b4b1c : SUCCESS in 53m 03s\n- puppet-openstack-integration-7-scenario005-tempest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86bfb06fb687497e958beee35c499fe0 : SUCCESS in 38m 27s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a2a48856cef4d88b16e6cb2850492c5 : SUCCESS in 56m 43s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/6285ed0c147f4b37a830bbaba169868b : SUCCESS in 12m 29s\n- puppet-openstack-litmus-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5521bad9da324c52af8a446d4985afb7 : SUCCESS in 4m 58s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e320c48a8cf4dd89424460805609a47 : SUCCESS in 4m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75a5014ec5e44b54823f73c507dcadfb : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":4},{"id":"3a42b49c770191fea66e0da4c602eb435b96f210","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2024-10-25 12:20:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9fe6e5e94d2dcb851d13b34490c2d0766be89406","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2024-10-25 13:53:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7d8aaff08a1cc5b5d1b3c8ec2d7d9a179916f04e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 13:53:55.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"3596ef1c6224307fa283109d878d9222013223d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 14:43:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0fb6e8b1a414253b369c195d341b41d\n\n- puppet-openstack-lint-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8d62a7f58994bc1aa903e82a29f2519 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/83012214bd69427fafec3112da4eeb4c : SUCCESS in 2m 11s\n- puppet-openstack-unit-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e5f72e49eea4d33b2cf939df098469e : SUCCESS in 7m 45s\n- puppet-openstack-unit-7.32-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/791e5cf244934010949b4e943dc8337f : SUCCESS in 3m 51s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5bc95c65e8d444668d47597af6269c0f : SUCCESS in 43m 42s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9b637fa55014315ad86a4a30272521d : SUCCESS in 39m 58s\n- puppet-openstack-litmus-7.32-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd755ffdbd9d4cc3af58f50639d2e16e : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90d19b1bd4c140199c37f35180ee64e5 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":4},{"id":"3d3be191d4e88740a92ad343eedd504fe715dedd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 14:44:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"fc0cb9c6dea3a7255082620882901b9e9185cd8c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 14:45:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/096657aba1bd4d0a91bfbceab6fb9cd6\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a363c235a0b84138ab0576b374687c9f : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7bdc37018d6c1f2169fdec6df53c45e9d83bdbda","revisions":{"209476746019e138012dfcbcee2126798ab7c6a1":{"kind":"REWORK","_number":1,"created":"2024-10-24 17:17:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/933340/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/40/933340/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/1"}}},"commit":{"parents":[{"commit":"ad75ab1eca25b5711d5a3995353c14b5110e18e1","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/ad75ab1eca25b5711d5a3995353c14b5110e18e1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 17:16:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 17:17:14.000000000","tz":540},"subject":"Support customizing inactivity probe for DB connections","message":"Support customizing inactivity probe for DB connections\n\nChange-Id: I88512683c88c403acca67d5d7c839d6fb7354557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/209476746019e138012dfcbcee2126798ab7c6a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/209476746019e138012dfcbcee2126798ab7c6a1"}]},"branch":"refs/heads/master"},"be6ebe70836ac1762f41aefe2dc2530a351cc0e1":{"kind":"REWORK","_number":2,"created":"2024-10-25 02:18:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/933340/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/40/933340/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/2"}}},"commit":{"parents":[{"commit":"ad75ab1eca25b5711d5a3995353c14b5110e18e1","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/ad75ab1eca25b5711d5a3995353c14b5110e18e1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 17:16:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-25 02:16:05.000000000","tz":540},"subject":"Support customizing inactivity probe for DB connections","message":"Support customizing inactivity probe for DB connections\n\nInactivity probe needs to be tuned in some deployments to avoid\nfrequent disconnection between neutron and OVN DBs. Expose parameters\nto tune the option for NB DB and SB DB.\n\nNote that configuring inactivity probe requires that connections are\nexplicitly set. So now connections are configured regardless of\nconnection protocol (tcp or ssl).\n\nChange-Id: I88512683c88c403acca67d5d7c839d6fb7354557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/be6ebe70836ac1762f41aefe2dc2530a351cc0e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/be6ebe70836ac1762f41aefe2dc2530a351cc0e1"}]},"branch":"refs/heads/master"},"e59793bc5799f649b71836819c07a117de5f2c01":{"kind":"REWORK","_number":3,"created":"2024-10-25 03:05:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/933340/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/40/933340/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/3"}}},"commit":{"parents":[{"commit":"ad75ab1eca25b5711d5a3995353c14b5110e18e1","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/ad75ab1eca25b5711d5a3995353c14b5110e18e1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 17:16:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-25 03:05:29.000000000","tz":540},"subject":"Support customizing inactivity probe for DB connections","message":"Support customizing inactivity probe for DB connections\n\nInactivity probe needs to be tuned in some deployments to avoid\nfrequent disconnection between neutron and OVN DBs. Expose parameters\nto tune the option for NB DB and SB DB.\n\nNote that configuring inactivity probe requires that connections are\nexplicitly set. So now connections are configured regardless of\nconnection protocol (tcp or ssl).\n\nChange-Id: I88512683c88c403acca67d5d7c839d6fb7354557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e59793bc5799f649b71836819c07a117de5f2c01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e59793bc5799f649b71836819c07a117de5f2c01"}]},"branch":"refs/heads/master"},"7bdc37018d6c1f2169fdec6df53c45e9d83bdbda":{"kind":"REWORK","_number":4,"created":"2024-10-25 08:57:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/933340/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/40/933340/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/40/933340/4"}}},"commit":{"parents":[{"commit":"ad75ab1eca25b5711d5a3995353c14b5110e18e1","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/ad75ab1eca25b5711d5a3995353c14b5110e18e1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 17:16:47.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-25 08:56:54.000000000","tz":540},"subject":"Support customizing inactivity probe for DB connections","message":"Support customizing inactivity probe for DB connections\n\nInactivity probe needs to be tuned in some deployments to avoid\nfrequent disconnection between neutron and OVN DBs. Expose parameters\nto tune the option for NB DB and SB DB.\n\nNote that configuring inactivity probe requires that connections are\nexplicitly set. So now connections are configured regardless of\nconnection protocol (tcp or ssl).\n\nChange-Id: I88512683c88c403acca67d5d7c839d6fb7354557\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/7bdc37018d6c1f2169fdec6df53c45e9d83bdbda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/7bdc37018d6c1f2169fdec6df53c45e9d83bdbda"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
