)]}'
{"id":"openstack%2Fpuppet-ovn~952047","triplet_id":"openstack%2Fpuppet-ovn~master~Ia75bf1eb5f527b99747069e07ba10d00241e9471","project":"openstack/puppet-ovn","branch":"master","attention_set":{},"removed_from_attention_set":{"37604":{"account":{"_account_id":37604,"name":"Richard Bachmann","display_name":"rbachman","email":"richard.bachmann@cern.ch","username":"rbachman","status":"Does cloud operations and network stuff at CERN."},"last_update":"2025-06-09 13:09:55.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-09 12:07:22.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ia75bf1eb5f527b99747069e07ba10d00241e9471","subject":"Put back northd refresh on config change.","status":"MERGED","created":"2025-06-08 14:23:10.000000000","updated":"2025-06-09 13:09:55.000000000","submitted":"2025-06-09 13:09:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952047","meta_rev_id":"a676eef0566be14ee310f63417678f5cb7c17b85","_number":952047,"virtual_id_number":952047,"owner":{"_account_id":37604,"name":"Richard Bachmann","display_name":"rbachman","email":"richard.bachmann@cern.ch","username":"rbachman","status":"Does cloud operations and network stuff at CERN."},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-09 13:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-06-09 12:07:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-06-09 12:20:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-06-09 12:20:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-08 14:58:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-09 11:57:50.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2025-06-09 11:57:50.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"fd9b41a3fb8781da9c7aa9dca990c72d7a363b3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37604,"name":"Richard Bachmann","display_name":"rbachman","email":"richard.bachmann@cern.ch","username":"rbachman","status":"Does cloud operations and network stuff at CERN."},"date":"2025-06-08 14:23:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"860e91fb9f3588ea94b8be9aa0cfa596c42f990d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-08 14:58:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9613ccf5fd90423ba0422e78fcaeec8d\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ad88040d65d142a48dd4e3bf783926b0 : SUCCESS in 1m 23s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d1bf98868f9c422990c37d530847c891 : SUCCESS in 2m 10s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c94b6a0fab6c45778666f3125e87b351 : SUCCESS in 1m 22s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/aaf5b6e5450a46888f60ed57b7304126 : SUCCESS in 5m 55s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e74d9e3f159747a3b5c7886cb93f672e : SUCCESS in 2m 14s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9213d78547604e49adbfb20394c06295 : SUCCESS in 3m 00s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/74c422788f0346edb2013ea2ad9b2d2c : SUCCESS in 21m 06s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fdf902051b94af189273ad410e82e86 : SUCCESS in 25m 57s\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5b45ffaa14f40729b48c8eb6f791588 : SUCCESS in 33m 52s (non-voting)\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/21dce022b0a84869a6d7aab3e34dc850 : SUCCESS in 26m 22s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/7abf115a4fa94892948dec131c25b1a0 : SUCCESS in 8m 34s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/195d7cf0fd424c1a93e37250353b89b8 : SUCCESS in 3m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/67b36b4735d147338dac8c8222652b69 : SUCCESS in 3m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3bf4fd8a38955f3a70311a7b4cba42c662689485","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-06-09 11:57:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1a8d2bf1b48df7df71f1298c59d449855247240","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-09 12:07:22.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ddecbd5f30ec36848b10f98dddcd9299b581adac","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-06-09 12:20:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45c171584bae8e9542a149f4d4917926293dc5e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 12:21:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7da4e7b324019c822a356cfdc9f693d4a5906f73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 13:09:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ff993d0dcd64f869a2ca8fd258384ee\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/aea7ba6fb31242d3b5a103ee26494a0d : SUCCESS in 2m 23s\n- puppet-openstack-syntax-7-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c356c15b4e10424d90f50c5595e915bb : SUCCESS in 2m 18s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d64b86d680b74f868aeb015a3999ac01 : SUCCESS in 2m 29s\n- puppet-openstack-unit-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cac0093728643fba97aa733e1384ae5 : SUCCESS in 6m 45s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4e7ef6cef15d4d1e8519a52e9edc414a : SUCCESS in 2m 14s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f01b915e0ab64be7b0e4bdf703dec8b6 : SUCCESS in 48m 22s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e7bbc3181314a0c807141423f441403 : SUCCESS in 46m 50s\n- puppet-openstack-litmus-7.34-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb821beea12c4451932a2b4d7009ac5a : SUCCESS in 10m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a676eef0566be14ee310f63417678f5cb7c17b85","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 13:09:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2eada288298bb1982b148144797191254f72f717","revisions":{"2eada288298bb1982b148144797191254f72f717":{"kind":"REWORK","_number":1,"created":"2025-06-08 14:23:10.000000000","uploader":{"_account_id":37604,"name":"Richard Bachmann","display_name":"rbachman","email":"richard.bachmann@cern.ch","username":"rbachman","status":"Does cloud operations and network stuff at CERN."},"ref":"refs/changes/47/952047/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/47/952047/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/47/952047/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/47/952047/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/47/952047/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/47/952047/1"}}},"commit":{"parents":[{"commit":"a4e29f41e7b1dd4d8c1089488ea098cad795ef06","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/a4e29f41e7b1dd4d8c1089488ea098cad795ef06"}]}],"author":{"name":"Richard Bachmann","email":"richard.bachmann@cern.ch","date":"2025-06-08 13:39:38.000000000","tz":120},"committer":{"name":"Richard Bachmann","email":"richard.bachmann@cern.ch","date":"2025-06-08 14:19:54.000000000","tz":120},"subject":"Put back northd refresh on config change.","message":"Put back northd refresh on config change.\n\nWhile adding a new node to an existing OVN cluster we ran into a small\nhiccup: The newly added node had its details correctly added to the\nconfig file at /etc/sysconfig/ovn-northd, but not to the parameters of\nthe ovn-northd process. The northd service was never refreshed.\n\nThis caused some issues down the line when a leadership change took\nplace and the new node became the leader in either of the\nnorth/southbound databases. At this point it could not connect to\nexisting northd nodes, resulting in an inability to update the DBs and\nthe OVN state going out of sync with regards to changes from Neutron.\n\nBy adding a refresh here we hope that the new node will automatically be\nadded to existing northd processes.\nI see that the inverse of this commit was introduced by\n66187619f2d781d1781f3a4f33c54ff82068b590, but I wonder if this was\nstrictly needed for that feature or incidental.\n\nChange-Id: Ia75bf1eb5f527b99747069e07ba10d00241e9471\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/2eada288298bb1982b148144797191254f72f717"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/2eada288298bb1982b148144797191254f72f717"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
