)]}'
{"id":"openstack%2Fpuppet-ovn~960952","triplet_id":"openstack%2Fpuppet-ovn~master~I0b1707c2bbe974d0fdf1b2cda5e44eff7276e4b3","project":"openstack/puppet-ovn","branch":"master","topic":"puppet-lint-rc","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-17 14:01:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0b1707c2bbe974d0fdf1b2cda5e44eff7276e4b3","subject":"Override disabled lint checks","status":"MERGED","created":"2025-09-13 08:04:04.000000000","updated":"2025-09-17 14:01:47.000000000","submitted":"2025-09-17 14:01:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960952-puppet-lint-rc","meta_rev_id":"5da73e7c8bc342cdaa466a2a9ca4eda1172c9e81","_number":960952,"virtual_id_number":960952,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-17 14:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2025-09-16 02:33:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2025-09-17 13:49:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2025-09-17 13:49:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-13 09:36:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-16 02:33:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2025-09-17 13:49:43.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"ef1a9cfd6e0093799b7f1dd49ce52dc2eb37b304","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-13 08:04:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93d4a5ae50a031af8d089341d9629054ef3d3135","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-13 08:19:37.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8a431eb7d1473d8aabeaf1acffaf055b94eeeaeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-13 08:20:19.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"b2fd8db8549626a463d597cd5d8c9009724d3b8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-13 09:31:25.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"54a6c69d5e0d0964952425eadc943a3c8a629f91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-13 09:36:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd44a9a76af54150aa657892e97a5169\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/848f37a8e7c94ff7b9770e0722054a64 : SUCCESS in 2m 29s","accounts_in_message":[],"_revision_number":4},{"id":"5548e43b839be52ff5aaa4fa80d480653e72e716","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-15 14:54:54.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"703bdc734685626ad2b91d8c3723be562888e6a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 15:10:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac5f5bd376b94e8e9b3117faf95c18f0\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0d900ecb7c954010a8bd849f1010b52d : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":5},{"id":"c8d5488c929c1a1460263fbeaba3a9bf400ee060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-15 15:35:18.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"9c155e1c7a750e10994e564490536ee7e3667771","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 15:42:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5bbe83465984afebf0a6b68d4164222\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/532d78e46702499eb3b65b3b1a98d368 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":6},{"id":"c55d30a8bcd4bf82fc74a30f694636b25b790486","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-15 15:44:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1bd48bbb13244b757c5f9f934c2512621c9599c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 15:54:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79240341268b46c0b7d5a5a26ae0b669\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/41dd32e048b9427aa7ea03a9a9dfff16 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":6},{"id":"40842eb70b664bd5c517257c6ef2bc2f677c9c2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-15 16:03:34.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"da2687b1e8c781bbff04914c569dc8435f516d70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 16:09:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eed2084bb3044dd81eedb091d99263c\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4827a54189054f539412f5c0af07dcc6 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":7},{"id":"3fda9db9e9d41978d04753c27616fe3032e2e53d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2025-09-16 02:33:47.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"1eb51d6e7f6eaf05f259b9744d961d034ef12d42","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-17 13:49:43.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1b9b013b0b816e7f09d8a08d47519575f05e0e12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 13:49:55.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"9364bfc1728119b019661a39c6ed9a7c5a668329","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 14:01:47.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee43ef4c22144ae7b6fd3772a2533331\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb88797ab9b44b0c841dc8c7995acc19 : SUCCESS in 2m 40s","accounts_in_message":[],"_revision_number":7},{"id":"5da73e7c8bc342cdaa466a2a9ca4eda1172c9e81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-17 14:01:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"00f54903b0640abf66cd5ec63d701d6aa01916d9","revisions":{"e5689e6ae772e3f9108e969a5454a8f1eeb028e7":{"kind":"REWORK","_number":1,"created":"2025-09-13 08:04:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/960952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/52/960952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/1"}}},"commit":{"parents":[{"commit":"9ea6d31b1ba022d3200f5fa0d15d16f94d06092c","subject":"Prepare for voxpupuli-puppet-lint-plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/9ea6d31b1ba022d3200f5fa0d15d16f94d06092c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-13 08:03:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-13 08:03:54.000000000","tz":540},"subject":"Maintain puppet-lint settings in individual repo","message":"Maintain puppet-lint settings in individual repo\n\n... so that we can turn on specific tests in a few modules.\n\nChange-Id: I0b1707c2bbe974d0fdf1b2cda5e44eff7276e4b3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e5689e6ae772e3f9108e969a5454a8f1eeb028e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e5689e6ae772e3f9108e969a5454a8f1eeb028e7"}]},"branch":"refs/heads/master"},"96ed4ab61db83d34c214bc14a051d229f2ecc54d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-09-13 08:19:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/960952/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/52/960952/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/2"}}},"commit":{"parents":[{"commit":"9ea6d31b1ba022d3200f5fa0d15d16f94d06092c","subject":"Prepare for voxpupuli-puppet-lint-plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/9ea6d31b1ba022d3200f5fa0d15d16f94d06092c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-13 08:03:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-13 08:19:37.000000000","tz":0},"subject":"Maintain puppet-lint settings in individual repo","message":"Maintain puppet-lint settings in individual repo\n\n... so that we can turn on specific tests in a few modules.\n\nNeeded-By: https://review.opendev.org/960932\nChange-Id: I0b1707c2bbe974d0fdf1b2cda5e44eff7276e4b3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/96ed4ab61db83d34c214bc14a051d229f2ecc54d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/96ed4ab61db83d34c214bc14a051d229f2ecc54d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7d6960cda9d516c06c3e280b3201ccdefcf7a6b1":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-09-13 08:20:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/960952/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/52/960952/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/3"}}},"commit":{"parents":[{"commit":"9ea6d31b1ba022d3200f5fa0d15d16f94d06092c","subject":"Prepare for voxpupuli-puppet-lint-plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/9ea6d31b1ba022d3200f5fa0d15d16f94d06092c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-13 08:03:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-13 08:20:19.000000000","tz":0},"subject":"Maintain puppet-lint settings in individual repo","message":"Maintain puppet-lint settings in individual repo\n\n... so that we can turn on specific tests in a few modules.\n\nDepends-on: https://review.opendev.org/960951\nNeeded-By: https://review.opendev.org/960932\nChange-Id: I0b1707c2bbe974d0fdf1b2cda5e44eff7276e4b3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/7d6960cda9d516c06c3e280b3201ccdefcf7a6b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/7d6960cda9d516c06c3e280b3201ccdefcf7a6b1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"931c70ae8dbf833adb31e399d892621cbf0c3a38":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-09-13 09:31:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/960952/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/52/960952/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/4"}}},"commit":{"parents":[{"commit":"9ea6d31b1ba022d3200f5fa0d15d16f94d06092c","subject":"Prepare for voxpupuli-puppet-lint-plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/9ea6d31b1ba022d3200f5fa0d15d16f94d06092c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-13 08:03:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-13 09:31:25.000000000","tz":0},"subject":"Maintain puppet-lint settings in individual repo","message":"Maintain puppet-lint settings in individual repo\n\n... so that we can turn on specific tests in a few modules.\n\nDepends-on: https://review.opendev.org/960960\nNeeded-By: https://review.opendev.org/960932\nChange-Id: I0b1707c2bbe974d0fdf1b2cda5e44eff7276e4b3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/931c70ae8dbf833adb31e399d892621cbf0c3a38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/931c70ae8dbf833adb31e399d892621cbf0c3a38"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ba657a1aa6c9dbbb805e0a7e9a0d9b1c10a56f18":{"kind":"REWORK","_number":5,"created":"2025-09-15 14:54:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/960952/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/52/960952/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/5"}}},"commit":{"parents":[{"commit":"9ea6d31b1ba022d3200f5fa0d15d16f94d06092c","subject":"Prepare for voxpupuli-puppet-lint-plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/9ea6d31b1ba022d3200f5fa0d15d16f94d06092c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-13 08:03:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-15 14:54:45.000000000","tz":540},"subject":"Maintain puppet-lint settings in individual repo","message":"Maintain puppet-lint settings in individual repo\n\n... so that we can turn on specific tests in a few modules.\n\nDepends-on: https://review.opendev.org/960960\nDepends-on: https://review.opendev.org/960932\nChange-Id: I0b1707c2bbe974d0fdf1b2cda5e44eff7276e4b3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/ba657a1aa6c9dbbb805e0a7e9a0d9b1c10a56f18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/ba657a1aa6c9dbbb805e0a7e9a0d9b1c10a56f18"}]},"branch":"refs/heads/master"},"e1e08f5b278d5251139cada3c7b81439ea4816a2":{"kind":"REWORK","_number":6,"created":"2025-09-15 15:35:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/960952/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/52/960952/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/6"}}},"commit":{"parents":[{"commit":"9ea6d31b1ba022d3200f5fa0d15d16f94d06092c","subject":"Prepare for voxpupuli-puppet-lint-plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/9ea6d31b1ba022d3200f5fa0d15d16f94d06092c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-13 08:03:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-15 15:35:07.000000000","tz":540},"subject":"Maintain puppet-lint settings in individual repo","message":"Maintain puppet-lint settings in individual repo\n\n... so that we can turn on specific tests in a few modules.\n\nDepends-on: https://review.opendev.org/960960\nDepends-on: https://review.opendev.org/960932\nChange-Id: I0b1707c2bbe974d0fdf1b2cda5e44eff7276e4b3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e1e08f5b278d5251139cada3c7b81439ea4816a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/e1e08f5b278d5251139cada3c7b81439ea4816a2"}]},"branch":"refs/heads/master"},"00f54903b0640abf66cd5ec63d701d6aa01916d9":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-09-15 16:03:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/960952/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/52/960952/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/52/960952/7"}}},"commit":{"parents":[{"commit":"9ea6d31b1ba022d3200f5fa0d15d16f94d06092c","subject":"Prepare for voxpupuli-puppet-lint-plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/9ea6d31b1ba022d3200f5fa0d15d16f94d06092c"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-13 08:03:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-15 16:03:34.000000000","tz":0},"subject":"Override disabled lint checks","message":"Override disabled lint checks\n\n... so that we can turn on a few more checks for this modules.\n\nDepends-on: https://review.opendev.org/960960\nDepends-on: https://review.opendev.org/960932\nChange-Id: I0b1707c2bbe974d0fdf1b2cda5e44eff7276e4b3\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/00f54903b0640abf66cd5ec63d701d6aa01916d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/00f54903b0640abf66cd5ec63d701d6aa01916d9"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
