)]}'
{"id":"openstack%2Fpuppet-ovn~980901","triplet_id":"openstack%2Fpuppet-ovn~master~I3bd21db98cdf9e8bfd1369a8aef4149f125fcfc4","project":"openstack/puppet-ovn","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-18 09:40:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3bd21db98cdf9e8bfd1369a8aef4149f125fcfc4","subject":"Ensure OVN DB services are up","status":"MERGED","created":"2026-03-17 11:29:05.000000000","updated":"2026-05-18 09:40:23.000000000","submitted":"2026-05-18 09:40:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"980901","meta_rev_id":"4ed4c11c388e53b7df04556036042882cbe3636a","_number":980901,"virtual_id_number":980901,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-18 09:40:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2026-05-11 19:25:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2026-05-18 08:20:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2026-05-18 08:20:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-17 12:38:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-05-11 19:25:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2026-05-18 08:20:59.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"7b7284581a19734bce97ef7f8879d6c6ee8fbd69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 11:29:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f248a1be38e52413ecbb306722dd236193053db0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 12:38:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01d205e3c29e48a892fa27ee0623888d\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/66bb6e705e2b4cac8497b29a4fa46dca : SUCCESS in 3m 07s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/59025628639b4fe3a798e63451eec473 : SUCCESS in 2m 20s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/27663dcaba7a4791b123e8901d20bda2 : SUCCESS in 4m 54s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9040d7a1235b4d53bc85fc9aa2540f54 : SUCCESS in 5m 39s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/805bc12373a84432a257a7ede7e4bcf6 : SUCCESS in 5m 42s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7a7b6d6e2518413489273ed9d02c2867 : FAILURE in 33m 20s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3dba421001a497ea668372793864bd4 : SUCCESS in 1h 02m 11s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b126a09e9b2f46dea6d372a472efff39 : SUCCESS in 59m 59s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f552d023ac354e299fb416015c62de58 : FAILURE in 32m 36s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cca1e41b16534dccbf4f553d7d4ed999 : SUCCESS in 32m 08s\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5238382c124249a78c53de1dee96796e : SUCCESS in 30m 21s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0d49fe785bc4f92a3f6d14c93ea22f1 : SUCCESS in 12m 23s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/abb149f686864dfdbd524d1af62554ba : SUCCESS in 4m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c58ab7be90bb45cc88fd771fa369b634 : FAILURE in 5m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"449aef77525862aeb2db2eac5bb0bbc016c9ec7b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 12:42:49.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"830de4df669e2ddd2d50f868327333634088d4f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-23 10:31:31.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"282392877c28621981d68c4f4d5aef553efd77a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-23 11:50:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb61188b3b6a4d4fbad0d6a647c3e8ba\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dd1856073cfc4bd6aec651faa33a5f1a : SUCCESS in 3m 38s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c172ccfb57074c719fc6a0567f301e86 : SUCCESS in 2m 54s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/c1f13a968aab4de79b18c07d338cfd41 : SUCCESS in 5m 41s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7402de488169493fb0bd5d6901696646 : SUCCESS in 6m 05s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ecb8b86f2a1f4248825834b1fdba8cd3 : SUCCESS in 6m 05s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b227ce82dd114cc9ac7eda99f93aaaec : SUCCESS in 58m 22s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/15bffd82657c49d28e6fed5faf8483fe : SUCCESS in 55m 52s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce43aa5b118042f8b74741f292e213b8 : SUCCESS in 35m 40s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfcf8292afe943968f54ec1f6371f43d : FAILURE in 51m 32s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e1ad9ebd8f34362b0c3dc759f396d74 : SUCCESS in 1h 05m 13s\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8dd45c444c648e1bb2466cb8fe13c1c : SUCCESS in 29m 48s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/67739c247355470c9095261b2bbc8328 : SUCCESS in 9m 45s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ebb53d1667a404e9b41be9b959330cc : SUCCESS in 7m 44s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/34202566e0734778837d2a71f72eee7a : SUCCESS in 6m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1fcc109e30bdae7e528183cf5c5e9f9142c5bd65","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-23 13:41:28.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"20c4bbc661aa9117a1ed3d5027caef025481935f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-23 13:41:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"21958de8d352e81861b2c508662cb1d34fcc943a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-24 13:11:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"83a2c46daf511deb1a9785e9b42877ef8404d426","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 14:21:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c47e2f8e3cf4519888becdcc58ea6cd\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3a4c8a5e043d4f9fa6e24fa80429c930 : POST_FAILURE in 2m 48s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f3b71fde12cc4d47b5ecce01d9c48ceb : SUCCESS in 2m 03s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/34ad616475c84398b4cd5b90bd0aa792 : POST_FAILURE in 4m 19s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c1f848cda4964fdbb9a86fcf020251b7 : POST_FAILURE in 3m 19s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2a6e287a9b374c6aaa285473cb145daf : POST_FAILURE in 5m 20s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cc2d5e5a3eb248c68d0d249103197a34 : SUCCESS in 55m 34s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d0057e36a974aad894ef554cf7199ec : POST_FAILURE in 46m 49s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/51400c8560794d36aace804932e77f34 : POST_FAILURE in 59m 54s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/065cc2e8ac2c4b78ae9a45c6144186b0 : FAILURE in 59m 47s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/209c7abcfee549b5a7ca5b0437ae72b7 : SUCCESS in 1h 03m 27s\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1da4ec2be58f40bcbad0742211e8f095 : SUCCESS in 58m 42s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/6312337a22744c14b3219e12793449d0 : POST_FAILURE in 12m 06s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5cd2d64afa49484686ee0e31009072bb : SUCCESS in 5m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fc1a70ce8032450cb3f242417d2a2e6d : SUCCESS in 5m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"72fa3d2bad1e2ca5ec01f6cd8e6d1036dc5580bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-09 14:33:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"72566ef5525cc72fe30b563b28ec292c294ff88c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-09 15:41:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/771b03e9cf404593b9ee030965156dd5\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/172fbf7434f449beb2923e810b2da193 : SUCCESS in 2m 03s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/163f83d424e94824b6657a4b40d4aae1 : SUCCESS in 2m 01s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/7a67cbd682aa4946b8669a5f7c98a7b6 : SUCCESS in 4m 55s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e682f257f2354c268ce852f388d88716 : SUCCESS in 4m 24s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e02a896f9b9241869a5a60771ed30025 : SUCCESS in 4m 22s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ce1ba08a97f454bb03f71f99e8e5091 : FAILURE in 34m 25s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/78597091114d486ba3ab30ff3d0ec685 : SUCCESS in 59m 13s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4367b8b5b59143d8b34896fac8686afc : SUCCESS in 1h 01m 36s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/792c79a7b9844546924b1aa828ae1bd9 : FAILURE in 15m 12s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/634409f2f07248a8a37069262e55483c : SUCCESS in 30m 38s\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/30252e2d37014c0085c820f912a67631 : SUCCESS in 57m 37s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/2015396d66b345cc9c8ed193cf2dcb2a : SUCCESS in 11m 00s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fa3bd5f8440649f4a917233d245e1876 : SUCCESS in 3m 51s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0766a084f9b64142872a842a263bdcc6 : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d8e857069c97a385e2a5fdba6a76de4ed3997743","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-10 01:13:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"77acd1ecc2d8f49acf1a71c10deb91af625b0add","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-10 02:28:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f91ad2c149b4b7e86097693212cb0df\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/efc8be3d7b73402aa8bba3bc4700ccb0 : SUCCESS in 2m 07s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/af5e53805393469396a0b9233796cac3 : SUCCESS in 3m 03s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/4a2ac764da56429daaa7df52c88c1201 : SUCCESS in 4m 36s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a5609ef2c9964072b6829d8ec5209a0b : SUCCESS in 4m 42s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1a43c75ad1064967b341ee243c988d19 : SUCCESS in 4m 25s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0634abd4ff3d481ea983029af03135bc : SUCCESS in 53m 03s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7c71eebc1cd42cab4385518995ad0f7 : SUCCESS in 1h 05m 49s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/559d1275f7404a78bde1c1177bd4313f : SUCCESS in 45m 07s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4fda6ddf610c45cfbfeaa2e9907b38fd : FAILURE in 58m 48s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c980c4648154e4fa21ca2e5908f542e : SUCCESS in 31m 57s\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/afc5b1894f27426ab2eb5faeb761ef6d : SUCCESS in 58m 17s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef22204bf086446499d96fe9f5370287 : SUCCESS in 7m 59s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d779ca3041f04befb2818df7a145b4a9 : SUCCESS in 6m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bdbabf794974f4ba4de6235a65294d8 : SUCCESS in 4m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"84a8a7ca11fc58ca48be79a4f037803713872e2f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-10 05:53:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"aea17f0c7fa53423e771c463851ca5e86c1a15a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-10 07:02:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dc4c0f76c5143ce9153da2353393702\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0cceafaaca824fb19139bbbf54be286a : SUCCESS in 2m 27s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/18f0b2c728d648a3b23b9650852a1c98 : SUCCESS in 3m 30s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/348093aefb194399b2a82270e9779d9e : SUCCESS in 4m 29s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3eae624853c344729cb1acf87314123f : SUCCESS in 4m 37s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6f31ad6a8c004ba1aa8810e479fa3ebb : SUCCESS in 5m 20s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/698467f4ebcc426bb9a32a05e6f7ed99 : SUCCESS in 37m 53s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/222018dbf7be45768649505481ca47ce : SUCCESS in 32m 53s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/52dd502042194d6ab07f22e2885b5b30 : SUCCESS in 1h 01m 29s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5d3a8699e003461fbc9400bfcee154e4 : FAILURE in 1h 00m 02s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/afc5bf73c621440b905dfbf5aa1c9673 : SUCCESS in 1h 02m 42s\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/798bdf7b578f47ab8fe5c66ea8b3d89c : SUCCESS in 1h 02m 20s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/343d085d8a0840c38abdb799ea9fdb62 : SUCCESS in 10m 39s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/936dc9c80c7b411b9eb0de1541ca8ac2 : SUCCESS in 3m 41s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7ff6bf0972f946dfb02898efb50bc72d : SUCCESS in 5m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"50379f8954afb4e9395a5d695ecbb226dcdbb74c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-05-10 13:47:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8d6eb760b8ba15635e58de077c1d6650a81de2ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-10 14:56:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2823a30b10c47f5892608e44dd7e43e\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1751739dec8d4dee86e9a6e688ae5612 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5a6ea6d8ab6041b2a8a4137217d3662d : SUCCESS in 2m 55s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/75ac69a098d049419b48cd6cbfb7340a : SUCCESS in 5m 44s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1bc3404139af494399b1f406842bee84 : SUCCESS in 4m 06s\n- puppet-openstack-unit-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a97a6da748d49ea9cf216b023307f0c : SUCCESS in 4m 29s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e26439d368934630ae86dd0df155791e : SUCCESS in 49m 18s (non-voting)\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b222eefe08e146288c4f6f0a46cda220 : SUCCESS in 1h 02m 25s\n- openvox-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/daddc30ebd694997a137e654a26353d5 : SUCCESS in 45m 27s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0b6b9552b40f4060ba93775b15ecd6bd : FAILURE in 55m 27s (non-voting)\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/daf4c7cb01cc4540a33e9c7cce8194f8 : SUCCESS in 56m 13s\n- openvox-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/27c4f64ca28549a7ac18030b32bde891 : SUCCESS in 58m 51s (non-voting)\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/56de56d5e191407890065452a2f69674 : SUCCESS in 13m 38s\n- puppet-openstack-litmus-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2eb27855e8824e3f9952cf557cdcf79c : SUCCESS in 5m 15s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb89bda3a1bf497294e48fa80ae49966 : SUCCESS in 5m 29s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7cdfcd8beebeb4f817e7eed3efbba2caf9b440d8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2026-05-11 19:25:13.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e708433e25d3ee8c1da5c1d53604538368046480","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2026-05-18 08:20:59.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"87dca13c7fc901648a05d92a621150f2aabf4e3f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 08:22:44.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6391da78ab5af6a4f301e09cd21cbcbec362fa24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 09:40:23.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b60b513a0564b4b9463c30e16c8e30a\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/38bafde3122849098e5e1f913c00c390 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-8-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5619ef324d98453285ce0684f1dea605 : SUCCESS in 3m 01s\n- puppet-openstack-unit-8.10-centos-10-stream https://zuul.opendev.org/t/openstack/build/92de38b074214e658c8f02784f7da797 : SUCCESS in 6m 57s\n- puppet-openstack-unit-8.10-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a807cde1225445f8bc185ba4c35bd5ce : SUCCESS in 4m 56s\n- puppet-openstack-integration-8-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/05008a2df2444644beb764c27ea23028 : SUCCESS in 55m 40s\n- puppet-openstack-integration-8-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1c6d73fe4304a2b851ba733e0e20ea1 : SUCCESS in 57m 37s\n- puppet-openstack-litmus-8.10-centos-9-stream https://zuul.opendev.org/t/openstack/build/49f1b4ff28914996be217a5847023dff : SUCCESS in 13m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b063978c875f4fc49f4dfde4f8e7107d : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":4},{"id":"4ed4c11c388e53b7df04556036042882cbe3636a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 09:40:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"71f2f5e7f8cdb73a2b9311e36bb1355b7cda3368","revisions":{"5721c8f7ca427efd2722592e2caa2732e1084b97":{"kind":"REWORK","_number":1,"created":"2026-03-17 11:29:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/980901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/01/980901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/1"}}},"commit":{"parents":[{"commit":"9cf18a2cc9f815a4f7a6fd7d09cd2da1e3b6cec0","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/9cf18a2cc9f815a4f7a6fd7d09cd2da1e3b6cec0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 11:27:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 11:28:31.000000000","tz":540},"subject":"Ensure OVN DB services are up","message":"Ensure OVN DB services are up\n\nIn Ubuntu we use ovn-central service to start all services, but this\ndoes not ensure that the actual services are started and fully ready.\nIntroduce the step to ensure the DB services are up, to avoid failure\nin exec resources which require access to these services.\n\nChange-Id: I3bd21db98cdf9e8bfd1369a8aef4149f125fcfc4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/5721c8f7ca427efd2722592e2caa2732e1084b97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/5721c8f7ca427efd2722592e2caa2732e1084b97"}]},"branch":"refs/heads/master"},"143f1e6e2939b1429fa8e1319d2ba843c4b3340f":{"kind":"REWORK","_number":2,"created":"2026-03-23 10:31:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/980901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/01/980901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/2"}}},"commit":{"parents":[{"commit":"9cf18a2cc9f815a4f7a6fd7d09cd2da1e3b6cec0","subject":"Update master for stable/2025.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/9cf18a2cc9f815a4f7a6fd7d09cd2da1e3b6cec0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 11:27:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-23 10:31:21.000000000","tz":540},"subject":"Ensure OVN DB services are up","message":"Ensure OVN DB services are up\n\nIn Ubuntu we use ovn-central service to start all services, but this\ndoes not ensure that the actual services are started and fully ready.\nIntroduce the step to ensure the DB services are up, to avoid failure\nin exec resources which require access to these services.\n\nChange-Id: I3bd21db98cdf9e8bfd1369a8aef4149f125fcfc4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/143f1e6e2939b1429fa8e1319d2ba843c4b3340f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/143f1e6e2939b1429fa8e1319d2ba843c4b3340f"}]},"branch":"refs/heads/master"},"7e97c7c9c8f4812b843ef289548b5430d1a3c513":{"kind":"REWORK","_number":3,"created":"2026-05-09 14:33:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/980901/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/01/980901/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/3"}}},"commit":{"parents":[{"commit":"72aa09f5c610880525dac34f939b542bdfe7c1b7","subject":"Prepare 2026.1 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/72aa09f5c610880525dac34f939b542bdfe7c1b7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 11:27:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-09 14:33:15.000000000","tz":540},"subject":"Ensure OVN DB services are up","message":"Ensure OVN DB services are up\n\nIn Ubuntu we use ovn-central service to start all services, but this\ndoes not ensure that the actual services are started and fully ready.\nIntroduce the step to ensure the DB services are up, to avoid failure\nin exec resources which require access to these services.\n\nChange-Id: I3bd21db98cdf9e8bfd1369a8aef4149f125fcfc4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/7e97c7c9c8f4812b843ef289548b5430d1a3c513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/7e97c7c9c8f4812b843ef289548b5430d1a3c513"}]},"branch":"refs/heads/master"},"71f2f5e7f8cdb73a2b9311e36bb1355b7cda3368":{"kind":"REWORK","_number":4,"created":"2026-05-10 01:13:05.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/980901/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-ovn","ref":"refs/changes/01/980901/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-ovn refs/changes/01/980901/4"}}},"commit":{"parents":[{"commit":"72aa09f5c610880525dac34f939b542bdfe7c1b7","subject":"Prepare 2026.1 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/72aa09f5c610880525dac34f939b542bdfe7c1b7"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 11:27:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-05-10 01:12:53.000000000","tz":540},"subject":"Ensure OVN DB services are up","message":"Ensure OVN DB services are up\n\nIn Ubuntu we use ovn-central service to start all services, but this\ndoes not ensure that the actual services are started and fully ready.\nIntroduce the step to ensure the DB services are up, to avoid failure\nin exec resources which require access to these services.\n\nChange-Id: I3bd21db98cdf9e8bfd1369a8aef4149f125fcfc4\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/71f2f5e7f8cdb73a2b9311e36bb1355b7cda3368"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-ovn/commit/71f2f5e7f8cdb73a2b9311e36bb1355b7cda3368"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
