)]}'
{"id":"openstack%2Fpuppet-pacemaker~660110","triplet_id":"openstack%2Fpuppet-pacemaker~master~I852d5d7aa8578c45f0c7215827cedd4ea72c8d0b","project":"openstack/puppet-pacemaker","branch":"master","topic":"pcs_newremote","hashtags":[],"change_id":"I852d5d7aa8578c45f0c7215827cedd4ea72c8d0b","subject":"Add new pcs 0.10 compatible remote addition implementation","status":"MERGED","created":"2019-05-20 12:56:32.000000000","updated":"2019-05-31 13:42:15.000000000","submitted":"2019-05-31 13:42:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660110-1559310135180-bed2ba93","meta_rev_id":"6398ad3d1516636f5d299584b4d9a46191f94832","_number":660110,"virtual_id_number":660110,"owner":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-05-31 13:42:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"date":"2019-05-31 08:16:01.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-05-31 13:25:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-05-31 13:25:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-23 06:05:49.000000000","updated_by":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"reviewer":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"state":"REVIEWER"},{"updated":"2019-05-30 16:42:45.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-05-31 08:16:01.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-31 13:25:28.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-05-31 13:42:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"619e75f8f3b07168bce4125f196fab9885004375","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-20 12:56:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dca2f5992b0ce5580e4452565a26a5d5f5c72727","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 13:22:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/660110/1/check/puppet-openstack-lint-ubuntu-bionic/31eb65b/ : SUCCESS in 4m 14s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/660110/1/check/puppet-openstack-syntax-4-ubuntu-bionic/3549060/ : FAILURE in 4m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/660110/1/check/puppet-openstack-syntax-5-ubuntu-bionic/4893f7f/ : FAILURE in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/660110/1/check/puppet-openstack-syntax-6-ubuntu-bionic/16980b4/ : FAILURE in 6m 49s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/660110/1/check/puppet-openstack-unit-4.8-centos-7/40371e3/ : SUCCESS in 10m 27s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/660110/1/check/puppet-openstack-unit-5.5-centos-7/11574b0/ : SUCCESS in 8m 40s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/10/660110/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/27f5008/ : SUCCESS in 7m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/10/660110/1/check/puppet-openstack-unit-latest-ubuntu-bionic/9858221/ : SUCCESS in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7e724e1cdf0c168abf95b0433a895f552afd285","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-20 13:26:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8faeb607477b67fc64ba39e1bfe75fec3c623a52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 14:00:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/660110/2/check/puppet-openstack-lint-ubuntu-bionic/5ee1d01/ : SUCCESS in 4m 18s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/660110/2/check/puppet-openstack-syntax-4-ubuntu-bionic/c4b1953/ : FAILURE in 4m 21s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/660110/2/check/puppet-openstack-syntax-5-ubuntu-bionic/069a5a2/ : FAILURE in 6m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/660110/2/check/puppet-openstack-syntax-6-ubuntu-bionic/208861a/ : FAILURE in 6m 57s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/660110/2/check/puppet-openstack-unit-4.8-centos-7/244f5a3/ : SUCCESS in 11m 17s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/660110/2/check/puppet-openstack-unit-5.5-centos-7/c38acd2/ : SUCCESS in 9m 53s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/10/660110/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/76346f5/ : SUCCESS in 7m 15s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/10/660110/2/check/puppet-openstack-unit-latest-ubuntu-bionic/fb760ea/ : SUCCESS in 12m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"62e093cb35d74741b3a4ac3bf9b69832b4a95b90","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-20 15:58:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/660110/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4e5d1b5/ : SUCCESS in 2h 25m 09s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/660110/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/b590703/ : SUCCESS in 2h 30m 02s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/10/660110/2/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/5a47b12/ : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":2},{"id":"2d8bfd9712e0070df80058347db6159564736f7d","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-21 10:39:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"21ae12f7d626532bb891bc3c589f1bb2525c53ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 11:13:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/660110/3/check/puppet-openstack-lint-ubuntu-bionic/ad26ce2/ : SUCCESS in 4m 42s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/660110/3/check/puppet-openstack-syntax-4-ubuntu-bionic/d918cbb/ : SUCCESS in 4m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/660110/3/check/puppet-openstack-syntax-5-ubuntu-bionic/4e8847d/ : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/660110/3/check/puppet-openstack-syntax-6-ubuntu-bionic/cffaa57/ : SUCCESS in 4m 52s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/660110/3/check/puppet-openstack-unit-4.8-centos-7/88a6cdd/ : SUCCESS in 10m 14s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/660110/3/check/puppet-openstack-unit-5.5-centos-7/e56dbbe/ : SUCCESS in 9m 39s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/10/660110/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/37c9665/ : SUCCESS in 7m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/10/660110/3/check/puppet-openstack-unit-latest-ubuntu-bionic/d6ea490/ : SUCCESS in 8m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b7ddfb88c82af6e7a9ac782a62eedeefb66a9e78","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-21 13:36:51.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/660110/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/4c8c153/ : FAILURE in 2h 55m 05s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/660110/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/bf728af/ : FAILURE in 2h 34m 38s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/10/660110/3/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/bbe347a/ : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":3},{"id":"210068516272bb7377f29dbe916c6e9aa0cf0ea6","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-23 05:55:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"13e19f41340b29971cebaa7aa5946679cbadb0ff","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-05-23 06:05:49.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d00f221dbb286d7d73ea90dfc0bd4b88d8d2372f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 06:38:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/660110/4/check/puppet-openstack-lint-ubuntu-bionic/5b31a6a/ : SUCCESS in 5m 31s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/660110/4/check/puppet-openstack-syntax-4-ubuntu-bionic/6031593/ : SUCCESS in 6m 32s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/660110/4/check/puppet-openstack-syntax-5-ubuntu-bionic/57af2c8/ : SUCCESS in 6m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/660110/4/check/puppet-openstack-syntax-6-ubuntu-bionic/85e0c4b/ : SUCCESS in 5m 26s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/660110/4/check/puppet-openstack-unit-4.8-centos-7/c9ebbec/ : SUCCESS in 9m 04s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/660110/4/check/puppet-openstack-unit-5.5-centos-7/3b7512a/ : SUCCESS in 11m 08s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/10/660110/4/check/puppet-openstack-unit-6.4-ubuntu-bionic/970fa04/ : SUCCESS in 7m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/10/660110/4/check/puppet-openstack-unit-latest-ubuntu-bionic/718cdac/ : SUCCESS in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"db5f2ebbc51a797f33530822009508f18319f443","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-23 06:53:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f732aa77a078bf851c9082ac509fa342fb3917ee","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 07:31:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/660110/5/check/puppet-openstack-lint-ubuntu-bionic/c30f9ba/ : SUCCESS in 4m 49s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/660110/5/check/puppet-openstack-syntax-4-ubuntu-bionic/a850281/ : SUCCESS in 6m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/660110/5/check/puppet-openstack-syntax-5-ubuntu-bionic/7fcf1e8/ : SUCCESS in 5m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/660110/5/check/puppet-openstack-syntax-6-ubuntu-bionic/3a6aacc/ : SUCCESS in 5m 30s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/660110/5/check/puppet-openstack-unit-4.8-centos-7/cc20c91/ : SUCCESS in 11m 04s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/660110/5/check/puppet-openstack-unit-5.5-centos-7/878fd07/ : SUCCESS in 9m 32s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/10/660110/5/check/puppet-openstack-unit-6.4-ubuntu-bionic/672a8ba/ : SUCCESS in 7m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/10/660110/5/check/puppet-openstack-unit-latest-ubuntu-bionic/febc7b3/ : SUCCESS in 10m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"76370ee0cbee7bc9a2b3b15f06ccd083c7d2f857","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-23 09:25:56.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/660110/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/c6b7886/ : SUCCESS in 2h 31m 50s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/660110/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/d6bdf61/ : FAILURE in 1h 27m 08s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/10/660110/5/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/4720ecf/ : SUCCESS in 1h 19m 38s","accounts_in_message":[],"_revision_number":5},{"id":"de4c3f19b37a089732e9bf753925f9530acd3bb3","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-30 12:56:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"85817fcbe7da5c00b30344cf59ba4e1493463322","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-30 13:19:21.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/660110/6/check/puppet-openstack-lint-ubuntu-bionic/f301fb0/ : SUCCESS in 4m 22s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/660110/6/check/puppet-openstack-syntax-4-ubuntu-bionic/7b40988/ : SUCCESS in 4m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/660110/6/check/puppet-openstack-syntax-5-ubuntu-bionic/26e55e0/ : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/660110/6/check/puppet-openstack-syntax-6-ubuntu-bionic/c7d1845/ : SUCCESS in 4m 43s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/660110/6/check/puppet-openstack-unit-4.8-centos-7/4859aa8/ : SUCCESS in 8m 48s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/660110/6/check/puppet-openstack-unit-5.5-centos-7/0769a90/ : SUCCESS in 9m 38s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/10/660110/6/check/puppet-openstack-unit-6.4-ubuntu-bionic/43985b6/ : SUCCESS in 8m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/10/660110/6/check/puppet-openstack-unit-latest-ubuntu-bionic/06cc9f5/ : SUCCESS in 6m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1f7eed7bfd48f93126570c3f08340a840e696118","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-30 15:20:04.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/660110/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/f82d5d1/ : SUCCESS in 2h 19m 26s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/660110/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/95628c7/ : SUCCESS in 2h 22m 02s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/10/660110/6/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/fe60e03/ : SUCCESS in 1h 10m 57s","accounts_in_message":[],"_revision_number":6},{"id":"75ac8d0614f1d48425fecc54891e12af07338bd5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-05-30 16:42:45.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7e596b59104befdf0bc3db14792f8e7d37e22bf1","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-31 05:43:55.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7b34e81de81d91f3a527bbb17648583e75c770f6","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-05-31 05:45:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f5e3c05a7379358a654c115f2691b93b545efe86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 06:00:32.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/660110/7/check/puppet-openstack-lint-ubuntu-bionic/41f0364/ : SUCCESS in 4m 11s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/660110/7/check/puppet-openstack-syntax-4-ubuntu-bionic/c885e01/ : SUCCESS in 4m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/660110/7/check/puppet-openstack-syntax-5-ubuntu-bionic/bfc20b5/ : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/660110/7/check/puppet-openstack-syntax-6-ubuntu-bionic/e48b053/ : SUCCESS in 4m 01s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/660110/7/check/puppet-openstack-unit-4.8-centos-7/0b90223/ : SUCCESS in 9m 36s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/660110/7/check/puppet-openstack-unit-5.5-centos-7/fd8c9ac/ : SUCCESS in 9m 01s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/10/660110/7/check/puppet-openstack-unit-6.4-ubuntu-bionic/1133d39/ : SUCCESS in 6m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/10/660110/7/check/puppet-openstack-unit-latest-ubuntu-bionic/12f7220/ : SUCCESS in 6m 47s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"4dbb0c6a059d409060eae12ed083f122d17bb22d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-05-31 08:16:01.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/10/660110/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/33e8c03/ : SUCCESS in 2h 29m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/10/660110/7/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/fac1fdf/ : SUCCESS in 2h 19m 02s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/10/660110/7/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/b6e6eb1/ : SUCCESS in 1h 13m 59s","accounts_in_message":[],"_revision_number":7},{"id":"ea6e4b3ee0925fdbf865a6ae7847c56647c5dc49","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-05-31 13:25:28.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"47a440f49fad109ff4ad58909d396fd86e4de66e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 13:25:46.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3198934af06bdef5bff4a2a1b77bf034be6440ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 13:42:14.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/10/660110/7/gate/puppet-openstack-lint-ubuntu-bionic/2f5b910/ : SUCCESS in 4m 26s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/10/660110/7/gate/puppet-openstack-syntax-4-ubuntu-bionic/e8240f4/ : SUCCESS in 4m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/10/660110/7/gate/puppet-openstack-syntax-5-ubuntu-bionic/5f22fc0/ : SUCCESS in 4m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/10/660110/7/gate/puppet-openstack-syntax-6-ubuntu-bionic/1eacdbb/ : SUCCESS in 4m 10s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/10/660110/7/gate/puppet-openstack-unit-4.8-centos-7/bd801e7/ : SUCCESS in 8m 44s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/10/660110/7/gate/puppet-openstack-unit-5.5-centos-7/fe995bb/ : SUCCESS in 10m 07s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/10/660110/7/gate/puppet-openstack-unit-6.4-ubuntu-bionic/27412b2/ : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":7},{"id":"dd507d775f95c753e14ccdc43b1a43b2674fb236","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 13:42:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"fde568564d43c78bf2828d3c981d5cc685fb3369","revisions":{"14506195dcf9eaccabbc861203950c0f4bf42ddf":{"kind":"REWORK","_number":1,"created":"2019-05-20 12:56:32.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/10/660110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/10/660110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/1"}}},"commit":{"parents":[{"commit":"a8921297aea2a139511af7e7f0c3f817fed783d1","subject":"Merge \"Do not avoid same node with INFINITY score\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/a8921297aea2a139511af7e7f0c3f817fed783d1"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-20 12:56:21.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-20 12:56:21.000000000","tz":120},"subject":"WIP pcs 0.10 new remote auth rework","message":"WIP pcs 0.10 new remote auth rework\n\nChange-Id: I852d5d7aa8578c45f0c7215827cedd4ea72c8d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/14506195dcf9eaccabbc861203950c0f4bf42ddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/14506195dcf9eaccabbc861203950c0f4bf42ddf"}]},"branch":"refs/heads/master"},"3a35e75355edcc1665be38274ef5b91ad9f55468":{"kind":"REWORK","_number":2,"created":"2019-05-20 13:26:35.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/10/660110/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/10/660110/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/2"}}},"commit":{"parents":[{"commit":"a8921297aea2a139511af7e7f0c3f817fed783d1","subject":"Merge \"Do not avoid same node with INFINITY score\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/a8921297aea2a139511af7e7f0c3f817fed783d1"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-20 12:56:21.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-20 13:26:25.000000000","tz":120},"subject":"WIP pcs 0.10 new remote auth rework","message":"WIP pcs 0.10 new remote auth rework\n\nChange-Id: I852d5d7aa8578c45f0c7215827cedd4ea72c8d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/3a35e75355edcc1665be38274ef5b91ad9f55468"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/3a35e75355edcc1665be38274ef5b91ad9f55468"}]},"branch":"refs/heads/master"},"04262ff46a3e039e921f9018d96fdffa21663429":{"kind":"REWORK","_number":3,"created":"2019-05-21 10:39:48.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/10/660110/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/10/660110/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/3"}}},"commit":{"parents":[{"commit":"a8921297aea2a139511af7e7f0c3f817fed783d1","subject":"Merge \"Do not avoid same node with INFINITY score\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/a8921297aea2a139511af7e7f0c3f817fed783d1"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-20 12:56:21.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-21 10:37:48.000000000","tz":120},"subject":"WIP pcs 0.10 new remote auth rework","message":"WIP pcs 0.10 new remote auth rework\n\nChange-Id: I852d5d7aa8578c45f0c7215827cedd4ea72c8d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/04262ff46a3e039e921f9018d96fdffa21663429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/04262ff46a3e039e921f9018d96fdffa21663429"}]},"branch":"refs/heads/master"},"632af6185d08a1bf55d21f74cb728c06da8c9ce9":{"kind":"REWORK","_number":4,"created":"2019-05-23 05:55:26.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/10/660110/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/10/660110/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/4"}}},"commit":{"parents":[{"commit":"0c5d69921dc61510109a5c53bbd963e6567f306c","subject":"Introduce a pcs_without_push function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/0c5d69921dc61510109a5c53bbd963e6567f306c"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-23 05:51:40.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-23 05:55:00.000000000","tz":120},"subject":"WIP Add new pcs 0.10 compatible remote addition implementation","message":"WIP Add new pcs 0.10 compatible remote addition implementation\n\nSmoketested this and correctly get remotes working on pcmk2.0/pcs\n0.10.x:\n[root@controller-0 galera-bundle-1]# pcs status\nCluster name: tripleo_cluster\nStack: corosync\nCurrent DC: controller-1 (version 2.0.1-4.el8-0eb7991564) - partition with quorum\nLast updated: Thu May 23 05:52:33 2019\nLast change: Tue May 21 14:20:14 2019 by root via cibadmin on controller-0\n\n17 nodes configured\n72 resources configured\n\nOnline: [ controller-0 controller-1 controller-2 ]\nRemoteOnline: [ compute-0 compute-1 ]\n\nFull list of resources:\n\n compute-0      (ocf::pacemaker:remote):        Started controller-0\n compute-1      (ocf::pacemaker:remote):        Started controller-1\n\nStill need to review the code one more time and then test the following:\nA) re-deploy idempotency\nB) Error cases\n\nChange-Id: I852d5d7aa8578c45f0c7215827cedd4ea72c8d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/632af6185d08a1bf55d21f74cb728c06da8c9ce9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/632af6185d08a1bf55d21f74cb728c06da8c9ce9"}]},"branch":"refs/heads/master"},"c4ae2803dd4cdfb950b88c422f504fe4b11f4490":{"kind":"REWORK","_number":5,"created":"2019-05-23 06:53:08.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/10/660110/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/10/660110/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/5"}}},"commit":{"parents":[{"commit":"0c5d69921dc61510109a5c53bbd963e6567f306c","subject":"Introduce a pcs_without_push function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/0c5d69921dc61510109a5c53bbd963e6567f306c"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-23 05:51:40.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-23 06:52:16.000000000","tz":120},"subject":"WIP Add new pcs 0.10 compatible remote addition implementation","message":"WIP Add new pcs 0.10 compatible remote addition implementation\n\nSmoketested this and correctly get remotes working on pcmk2.0/pcs\n0.10.x:\n[root@controller-0 galera-bundle-1]# pcs status\nCluster name: tripleo_cluster\nStack: corosync\nCurrent DC: controller-1 (version 2.0.1-4.el8-0eb7991564) - partition with quorum\nLast updated: Thu May 23 05:52:33 2019\nLast change: Tue May 21 14:20:14 2019 by root via cibadmin on controller-0\n\n17 nodes configured\n72 resources configured\n\nOnline: [ controller-0 controller-1 controller-2 ]\nRemoteOnline: [ compute-0 compute-1 ]\n\nFull list of resources:\n\n compute-0      (ocf::pacemaker:remote):        Started controller-0\n compute-1      (ocf::pacemaker:remote):        Started controller-1\n\nStill need to review the code one more time and then test the following:\nA) re-deploy idempotency\nB) Error cases\n\nChange-Id: I852d5d7aa8578c45f0c7215827cedd4ea72c8d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/c4ae2803dd4cdfb950b88c422f504fe4b11f4490"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/c4ae2803dd4cdfb950b88c422f504fe4b11f4490"}]},"branch":"refs/heads/master"},"986388d3eea33dfd419c74ccddef16ad5d86634f":{"kind":"REWORK","_number":6,"created":"2019-05-30 12:56:26.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/10/660110/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/10/660110/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/6"}}},"commit":{"parents":[{"commit":"0c5d69921dc61510109a5c53bbd963e6567f306c","subject":"Introduce a pcs_without_push function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/0c5d69921dc61510109a5c53bbd963e6567f306c"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-23 05:51:40.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-30 12:50:16.000000000","tz":120},"subject":"Add new pcs 0.10 compatible remote addition implementation","message":"Add new pcs 0.10 compatible remote addition implementation\n\nNew pcs 0.10 needs to implement remotes addition in a completely\ndifferent way. It is not sufficient anymore to just create the\nocf::pacemaker::remote resource and make sure that pacemaker_remote is\nup and running with the same authkey on all nodes. Now we\nneed to setup pcsd on all nodes (remotes and controllers), setup the\nhacluster user and then do a pcs cluster remote-add command.\n\nWe do this only when pcs 0.10 is detected in order to maintain\nthe old proven behaviour on pcs 0.9x.\n\nTested this on stein + rhel8 correctly get remotes working on an IHA\nsetup:\n[root@controller-0 galera-bundle-1]# pcs status\nCluster name: tripleo_cluster\nStack: corosync\nCurrent DC: controller-1 (version 2.0.1-4.el8-0eb7991564) - partition with quorum\nLast updated: Thu May 23 05:52:33 2019\nLast change: Tue May 21 14:20:14 2019 by root via cibadmin on controller-0\n\n17 nodes configured\n72 resources configured\n\nOnline: [ controller-0 controller-1 controller-2 ]\nRemoteOnline: [ compute-0 compute-1 ]\n\nFull list of resources:\n\n compute-0      (ocf::pacemaker:remote):        Started controller-0\n compute-1      (ocf::pacemaker:remote):        Started controller-1\n\nAlso tested on qiqueens SP13 to make sure no regressions in the old codepaths\nare introduced. Additionally tested a redeploy on stein to make sure\nidempotency is preserved.\n\nChange-Id: I852d5d7aa8578c45f0c7215827cedd4ea72c8d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/986388d3eea33dfd419c74ccddef16ad5d86634f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/986388d3eea33dfd419c74ccddef16ad5d86634f"}]},"branch":"refs/heads/master"},"fde568564d43c78bf2828d3c981d5cc685fb3369":{"kind":"REWORK","_number":7,"created":"2019-05-31 05:45:08.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/10/660110/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/10/660110/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/10/660110/7"}}},"commit":{"parents":[{"commit":"0c5d69921dc61510109a5c53bbd963e6567f306c","subject":"Introduce a pcs_without_push function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/0c5d69921dc61510109a5c53bbd963e6567f306c"}]}],"author":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-23 05:51:40.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-05-31 05:44:55.000000000","tz":120},"subject":"Add new pcs 0.10 compatible remote addition implementation","message":"Add new pcs 0.10 compatible remote addition implementation\n\nNew pcs 0.10 needs to implement remotes addition in a completely\ndifferent way. It is not sufficient anymore to just create the\nocf::pacemaker::remote resource and make sure that pacemaker_remote is\nup and running with the same authkey on all nodes. Now we\nneed to setup pcsd on all nodes (remotes and controllers), setup the\nhacluster user and then do a pcs cluster remote-add command.\n\nWe do this only when pcs 0.10 is detected in order to maintain\nthe old proven behaviour on pcs 0.9x.\n\nTested this on stein + rhel8 correctly get remotes working on an IHA\nsetup:\n[root@controller-0 galera-bundle-1]# pcs status\nCluster name: tripleo_cluster\nStack: corosync\nCurrent DC: controller-1 (version 2.0.1-4.el8-0eb7991564) - partition with quorum\nLast updated: Thu May 23 05:52:33 2019\nLast change: Tue May 21 14:20:14 2019 by root via cibadmin on controller-0\n\n17 nodes configured\n72 resources configured\n\nOnline: [ controller-0 controller-1 controller-2 ]\nRemoteOnline: [ compute-0 compute-1 ]\n\nFull list of resources:\n\n compute-0      (ocf::pacemaker:remote):        Started controller-0\n compute-1      (ocf::pacemaker:remote):        Started controller-1\n\nAlso tested on queens to make sure no regressions in the old codepaths\nare introduced. Additionally tested a redeploy on stein to make sure\nidempotency is preserved.\n\nChange-Id: I852d5d7aa8578c45f0c7215827cedd4ea72c8d0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/fde568564d43c78bf2828d3c981d5cc685fb3369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/fde568564d43c78bf2828d3c981d5cc685fb3369"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
