)]}'
{"id":"openstack%2Fpuppet-pacemaker~662740","triplet_id":"openstack%2Fpuppet-pacemaker~master~Iad1193c1df0792dcb7dce2af3575e19e58781ba0","project":"openstack/puppet-pacemaker","branch":"master","hashtags":[],"change_id":"Iad1193c1df0792dcb7dce2af3575e19e58781ba0","subject":"Support for SBD (watchdog only)","status":"MERGED","created":"2019-06-03 12:23:02.000000000","updated":"2019-08-05 13:03:55.000000000","submitted":"2019-08-05 13:03:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662740-1565010235419-c33df9d0","meta_rev_id":"7281096178f5674d0520308a16c31173eb9c7df6","_number":662740,"virtual_id_number":662740,"owner":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-08-05 13:03:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"date":"2019-08-02 19:47:30.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-08-02 16:40:48.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-02 16:40:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-08-05 12:46:08.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"value":0,"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-24 06:07:57.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-08-02 13:26:17.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-08-02 19:47:30.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-05 12:46:08.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-08-05 13:03:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ad9e2481e02298829bf18ca5a03f2e199762ec3b","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-06-03 12:23:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f0cd4b160848185c049c1bc09106afb4dda6609","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-06-03 12:24:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e7bf664b2388c5b0efa03d0098c3f51750c479ca","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-06-03 12:36:49.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2b7e1850693bb191ec04cfe7ec2e34951ade8a05","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 12:52:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/40/662740/3/check/puppet-openstack-lint-ubuntu-bionic/a0533cf/ : SUCCESS in 4m 06s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/40/662740/3/check/puppet-openstack-syntax-4-ubuntu-bionic/b80057f/ : SUCCESS in 6m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/40/662740/3/check/puppet-openstack-syntax-5-ubuntu-bionic/6c2c49a/ : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/40/662740/3/check/puppet-openstack-syntax-6-ubuntu-bionic/0e7f7ea/ : SUCCESS in 4m 06s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/40/662740/3/check/puppet-openstack-unit-4.8-centos-7/0cb8d48/ : SUCCESS in 9m 42s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/40/662740/3/check/puppet-openstack-unit-5.5-centos-7/978ad3f/ : SUCCESS in 9m 06s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/40/662740/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/4ee7c6e/ : SUCCESS in 6m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/40/662740/3/check/puppet-openstack-unit-latest-ubuntu-bionic/659bb20/ : SUCCESS in 6m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dcd58419401f6483752196f47c29a0c9b2db6c32","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-06-03 13:20:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d2bf2113f305b7b2d6ca2680a2b6d0afb46309b1","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-06-03 14:50:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"37eb1ec5b6473a6d36d4bf8120f963106e9c5291","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-06-03 14:52:15.000000000","message":"Patch Set 4:\n\n[heat-admin@controller-0 ~]$ sudo pcs status\nCluster name: tripleo_cluster\nStack: corosync\nCurrent DC: controller-2 (version 2.0.1-4.el8-0eb7991564) - partition with quorum\nLast updated: Mon Jun  3 14:51:28 2019\nLast change: Mon Jun  3 14:47:41 2019 by root via cibadmin on controller-0\n\n3 nodes configured\n0 resources configured\n\nOnline: [ controller-0 controller-1 controller-2 ]\n\nNo resources\n\n\nDaemon Status:\n  corosync: active/enabled\n  pacemaker: active/enabled\n  pcsd: active/enabled\n  sbd: active/enabled\n\n\n[heat-admin@controller-0 ~]$ sudo pcs stonith sbd config\nSBD_DELAY_START\u003dno\nSBD_STARTMODE\u003dalways\nSBD_WATCHDOG_TIMEOUT\u003d20\n\nWatchdogs:\n  controller-1: /dev/watchdog\n  controller-0: /dev/watchdog\n  controller-2: /dev/watchdog\n\n[heat-admin@controller-0 ~]$ sudo pcs stonith sbd status\nSBD STATUS\n\u003cnode name\u003e: \u003cinstalled\u003e | \u003cenabled\u003e | \u003crunning\u003e\ncontroller-0: YES | YES | YES\ncontroller-1: YES | YES | YES\ncontroller-2: YES | YES | YES","accounts_in_message":[],"_revision_number":4},{"id":"71ad659460283452e1ac3f501213a56418d4fd2d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 15:22:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/40/662740/4/check/puppet-openstack-lint-ubuntu-bionic/35ac599/ : SUCCESS in 4m 29s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/40/662740/4/check/puppet-openstack-syntax-4-ubuntu-bionic/4c16347/ : SUCCESS in 4m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/40/662740/4/check/puppet-openstack-syntax-5-ubuntu-bionic/f83a860/ : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/40/662740/4/check/puppet-openstack-syntax-6-ubuntu-bionic/f79101a/ : RETRY_LIMIT in 1m 14s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/40/662740/4/check/puppet-openstack-unit-4.8-centos-7/098b726/ : SUCCESS in 9m 43s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/40/662740/4/check/puppet-openstack-unit-5.5-centos-7/cd62272/ : SUCCESS in 9m 54s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/40/662740/4/check/puppet-openstack-unit-6.4-ubuntu-bionic/48ef86f/ : SUCCESS in 8m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/40/662740/4/check/puppet-openstack-unit-latest-ubuntu-bionic/d80951c/ : SUCCESS in 8m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"325f708c0c2132c343bed5a536359c3eefe18094","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-06-03 15:56:29.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"05b97a968c9928c942149c4ccbee637f6c909085","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 16:15:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/40/662740/4/check/puppet-openstack-lint-ubuntu-bionic/02e0912/ : SUCCESS in 5m 35s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/40/662740/4/check/puppet-openstack-syntax-4-ubuntu-bionic/7a5b960/ : SUCCESS in 4m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/40/662740/4/check/puppet-openstack-syntax-5-ubuntu-bionic/c764a0d/ : SUCCESS in 4m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/40/662740/4/check/puppet-openstack-syntax-6-ubuntu-bionic/4544c88/ : SUCCESS in 4m 09s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/40/662740/4/check/puppet-openstack-unit-4.8-centos-7/48fe8e7/ : SUCCESS in 10m 07s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/40/662740/4/check/puppet-openstack-unit-5.5-centos-7/fbfb9e6/ : SUCCESS in 9m 28s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/40/662740/4/check/puppet-openstack-unit-6.4-ubuntu-bionic/4b239e3/ : SUCCESS in 6m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/40/662740/4/check/puppet-openstack-unit-latest-ubuntu-bionic/72dbc30/ : SUCCESS in 7m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4d2737bf3febde3f6c3527da44d984c3c4381ee1","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-06-03 17:33:59.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/40/662740/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/67a5b92/ : FAILURE in 1h 04m 35s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/40/662740/4/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/59c8b0e/ : SUCCESS in 2h 27m 50s\n- tripleo-ci-centos-7-multinode-1ctlr-featureset010 http://logs.rdoproject.org/40/662740/4/openstack-check/tripleo-ci-centos-7-multinode-1ctlr-featureset010/e3f25cc/ : SUCCESS in 1h 12m 39s","accounts_in_message":[],"_revision_number":4},{"id":"2681e4b27e7928b4a50e329d7c844cde7a99cded","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-07-24 05:44:45.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"43ff5e59a69d4bbf8d6a3b6453567911a586661e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 05:59:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/40/662740/5/check/puppet-openstack-lint-ubuntu-bionic/bf00696/ : SUCCESS in 4m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/40/662740/5/check/puppet-openstack-syntax-5-ubuntu-bionic/7bfd5bb/ : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/40/662740/5/check/puppet-openstack-syntax-6-ubuntu-bionic/98834df/ : SUCCESS in 4m 46s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/40/662740/5/check/puppet-openstack-unit-5.5-centos-7/cd2b847/ : SUCCESS in 9m 41s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/40/662740/5/check/puppet-openstack-unit-6.4-ubuntu-bionic/e7e104e/ : SUCCESS in 7m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/40/662740/5/check/puppet-openstack-unit-latest-ubuntu-bionic/e5eb3f1/ : SUCCESS in 7m 37s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"d4d2d5dbcfe48372414575229cc60f76cd19368a","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-24 06:07:57.000000000","message":"Patch Set 5: Code-Review+1\n\nlgtm","accounts_in_message":[],"_revision_number":5},{"id":"33b6fb86f9d36d8b0efc503d9f696b7a0b7fac7d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-24 08:29:35.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/40/662740/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/d8e0b01/ : SUCCESS in 2h 42m 13s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/40/662740/5/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/9fc4538/ : SUCCESS in 2h 34m 15s","accounts_in_message":[],"_revision_number":5},{"id":"92fa4a7850e6afc87505bb951bceb5852eb87763","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-29 17:16:45.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7b4507560759a2efdaf0d16230130abfe7e84470","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-31 20:17:25.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"32b3a69e2101a39faf466e0dbfd3030e26ff8f5f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-01 12:23:41.000000000","message":"Patch Set 5: -Code-Review","accounts_in_message":[],"_revision_number":5},{"id":"dd0ce8c34ca342c108d6e1c91a3ac934aab72e34","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-08-02 08:33:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0132c3b764bd3e52f31da84a96fbc86ed91646e2","author":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"date":"2019-08-02 08:37:40.000000000","message":"Patch Set 6: Code-Review+1\n\n\\o/","accounts_in_message":[],"_revision_number":6},{"id":"3ef611329a91fee3a0ff73bfb8efcc3ff0f357ab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 08:50:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/40/662740/6/check/puppet-openstack-lint-ubuntu-bionic/3de8c1e/ : SUCCESS in 4m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/40/662740/6/check/puppet-openstack-syntax-5-ubuntu-bionic/815400a/ : SUCCESS in 4m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/40/662740/6/check/puppet-openstack-syntax-6-ubuntu-bionic/680301a/ : SUCCESS in 6m 51s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/40/662740/6/check/puppet-openstack-unit-5.5-centos-7/12f3f82/ : SUCCESS in 9m 48s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://logs.opendev.org/40/662740/6/check/puppet-openstack-unit-6.4-ubuntu-bionic/8945f9c/ : SUCCESS in 11m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/40/662740/6/check/puppet-openstack-unit-latest-ubuntu-bionic/51400b2/ : SUCCESS in 10m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ced7a78bbf526513a24c6a22908b859416ea7373","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 12:33:00.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/40/662740/6/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/18eede1/ : FAILURE in 1h 59m 29s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 finger://ze07.softwarefactory-project.io:7979/c8c55ee341284e3bb4f700989289dc57 : POST_FAILURE in 3h 57m 42s","accounts_in_message":[],"_revision_number":6},{"id":"0a32dcbdf322fc4fed6ee48c78067864c364e663","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-02 13:26:17.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"2e8295abb944e5e73f230da0e0199a82fe76168f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-02 15:49:00.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"3f4502780f3ff111de0bd25d002e74291b961375","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 16:03:49.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/40/662740/6/check/puppet-openstack-lint-ubuntu-bionic/42556d2/ : SUCCESS in 4m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/40/662740/6/check/puppet-openstack-syntax-5-ubuntu-bionic/c144f6d/ : SUCCESS in 5m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/40/662740/6/check/puppet-openstack-syntax-6-ubuntu-bionic/ec73d48/ : SUCCESS in 3m 58s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/40/662740/6/check/puppet-openstack-unit-5.5-centos-7/fc97ac7/ : FAILURE in 6m 40s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://logs.opendev.org/40/662740/6/check/puppet-openstack-unit-6.4-ubuntu-bionic/ff0cfbf/ : SUCCESS in 10m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/40/662740/6/check/puppet-openstack-unit-latest-ubuntu-bionic/7dda6a7/ : SUCCESS in 10m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"45dda00dbe85dd52f1df699fa6b6cd4e491091e7","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-02 16:12:40.000000000","message":"Patch Set 6:\n\npuppet-pacemaker doesn\u0027t use the openstack spec helper so it needs to pin correctly","accounts_in_message":[],"_revision_number":6},{"id":"e55369eb3338ec8cfaa713cd8ddb1aeaf38806f5","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-02 16:22:03.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"893980dbde2599360559d5ac49741329394b5a97","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 16:37:02.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/40/662740/7/check/puppet-openstack-lint-ubuntu-bionic/4cd3c82/ : FAILURE in 3m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/40/662740/7/check/puppet-openstack-syntax-5-ubuntu-bionic/98ad1dc/ : FAILURE in 2m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/40/662740/7/check/puppet-openstack-syntax-6-ubuntu-bionic/b77d0b1/ : FAILURE in 3m 52s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/40/662740/7/check/puppet-openstack-unit-5.5-centos-7/c33055f/ : FAILURE in 7m 37s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://logs.opendev.org/40/662740/7/check/puppet-openstack-unit-6.4-ubuntu-bionic/0a90902/ : FAILURE in 5m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/40/662740/7/check/puppet-openstack-unit-latest-ubuntu-bionic/f36be63/ : FAILURE in 5m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0f2cb5821a0909495082178ce566ab5b20e7643f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-08-02 16:40:48.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"2ec315403eab28e7a366992f3ee0729631046853","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-02 16:58:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/40/662740/8/check/puppet-openstack-lint-ubuntu-bionic/4dbf280/ : SUCCESS in 5m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/40/662740/8/check/puppet-openstack-syntax-5-ubuntu-bionic/641e605/ : SUCCESS in 5m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/40/662740/8/check/puppet-openstack-syntax-6-ubuntu-bionic/5e98652/ : SUCCESS in 4m 27s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/40/662740/8/check/puppet-openstack-unit-5.5-centos-7/17c5362/ : SUCCESS in 11m 31s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://logs.opendev.org/40/662740/8/check/puppet-openstack-unit-6.4-ubuntu-bionic/e0f156c/ : SUCCESS in 10m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://logs.opendev.org/40/662740/8/check/puppet-openstack-unit-latest-ubuntu-bionic/ee0a6e2/ : SUCCESS in 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"69b337a1bafbe44b4cb32d4a843419dc643484f3","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-08-02 19:47:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/40/662740/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/089f957/ : SUCCESS in 3h 05m 54s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/40/662740/8/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/ccb8e6d/ : SUCCESS in 2h 44m 02s","accounts_in_message":[],"_revision_number":8},{"id":"0574bc65b61978596bf9517eca68226e7ce7c3bb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-08-05 12:46:08.000000000","message":"Patch Set 8: Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"030895425b6fb13759e59d1272b80fdf2f592875","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 12:46:19.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"32fdb2ac321646e41f7e9374cbb4e1048786bb27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 13:03:55.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://logs.opendev.org/40/662740/8/gate/puppet-openstack-lint-ubuntu-bionic/1ff6330/ : SUCCESS in 5m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://logs.opendev.org/40/662740/8/gate/puppet-openstack-syntax-5-ubuntu-bionic/d345f81/ : SUCCESS in 5m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://logs.opendev.org/40/662740/8/gate/puppet-openstack-syntax-6-ubuntu-bionic/a50dbbe/ : SUCCESS in 5m 06s\n- puppet-openstack-unit-5.5-centos-7 https://logs.opendev.org/40/662740/8/gate/puppet-openstack-unit-5.5-centos-7/146fb9a/ : SUCCESS in 11m 52s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://logs.opendev.org/40/662740/8/gate/puppet-openstack-unit-6.4-ubuntu-bionic/5329979/ : SUCCESS in 9m 27s","accounts_in_message":[],"_revision_number":8},{"id":"7281096178f5674d0520308a16c31173eb9c7df6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-05 13:03:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"519c28b163cef8ef21e83d7c177c0936de55e183","revisions":{"c301ee79a43d6d4734e6b6108d35ce817b7453da":{"kind":"REWORK","_number":1,"created":"2019-06-03 12:23:02.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/40/662740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/40/662740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/1"}}},"commit":{"parents":[{"commit":"fde568564d43c78bf2828d3c981d5cc685fb3369","subject":"Add new pcs 0.10 compatible remote addition implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/fde568564d43c78bf2828d3c981d5cc685fb3369"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-06-03 12:17:55.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-06-03 12:23:00.000000000","tz":120},"subject":"WIP - Support for SBD (watchdog only)","message":"WIP - Support for SBD (watchdog only)\n\nThis commit allow the operators to configure SBD-based fencing.\n\nBecause enabling sbd requires a cluster restart the assumption\nis that sbd will be enabled at cluster creation time by setting\n\"pacemaker::corosync::enable_sbd\" to true.\n\nparameter_defaults:\n  ExtraConfig:\n    pacemaker::corosync::enable_sbd: true\n\nAn optional watchdog_timeout value (in seconds) can be also specified via:\n\n    pacemaker::corosync::sbd_watchdog_timeout: 20\n\nIf unset, the default value will be set to 10s.\n\nChange-Id: Iad1193c1df0792dcb7dce2af3575e19e58781ba0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/c301ee79a43d6d4734e6b6108d35ce817b7453da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/c301ee79a43d6d4734e6b6108d35ce817b7453da"}]},"branch":"refs/heads/master"},"ebd12c95a5cbd891c362de5750e2c7cc8a7c1341":{"kind":"REWORK","_number":2,"created":"2019-06-03 12:24:59.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/40/662740/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/40/662740/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/2"}}},"commit":{"parents":[{"commit":"fde568564d43c78bf2828d3c981d5cc685fb3369","subject":"Add new pcs 0.10 compatible remote addition implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/fde568564d43c78bf2828d3c981d5cc685fb3369"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-06-03 12:17:55.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-06-03 12:24:52.000000000","tz":120},"subject":"WIP - Support for SBD (watchdog only)","message":"WIP - Support for SBD (watchdog only)\n\nThis commit allow the operators to configure SBD-based fencing.\n\nBecause enabling sbd requires a cluster restart the assumption\nis that sbd will be enabled at cluster creation time by setting\n\"pacemaker::corosync::enable_sbd\" to true.\n\nparameter_defaults:\n  ExtraConfig:\n    pacemaker::corosync::enable_sbd: true\n\nAn optional watchdog_timeout value (in seconds) can be also specified via:\n\n    pacemaker::corosync::sbd_watchdog_timeout: 20\n\nIf unset, the default value will be set to 10s.\n\nChange-Id: Iad1193c1df0792dcb7dce2af3575e19e58781ba0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/ebd12c95a5cbd891c362de5750e2c7cc8a7c1341"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/ebd12c95a5cbd891c362de5750e2c7cc8a7c1341"}]},"branch":"refs/heads/master"},"ea1d82876704575f3a48125b30cb224e3b2ad3a3":{"kind":"REWORK","_number":3,"created":"2019-06-03 12:36:49.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/40/662740/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/40/662740/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/3"}}},"commit":{"parents":[{"commit":"fde568564d43c78bf2828d3c981d5cc685fb3369","subject":"Add new pcs 0.10 compatible remote addition implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/fde568564d43c78bf2828d3c981d5cc685fb3369"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-06-03 12:17:55.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-06-03 12:36:41.000000000","tz":120},"subject":"WIP - Support for SBD (watchdog only)","message":"WIP - Support for SBD (watchdog only)\n\nThis commit allow the operators to configure SBD-based fencing.\n\nBecause enabling sbd requires a cluster restart the assumption\nis that sbd will be enabled at cluster creation time by setting\n\"pacemaker::corosync::enable_sbd\" to true.\n\nparameter_defaults:\n  ExtraConfig:\n    pacemaker::corosync::enable_sbd: true\n\nAn optional watchdog_timeout value (in seconds) can be also specified via:\n\n    pacemaker::corosync::sbd_watchdog_timeout: 20\n\nIf unset, the default value will be set to 10s.\n\nChange-Id: Iad1193c1df0792dcb7dce2af3575e19e58781ba0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/ea1d82876704575f3a48125b30cb224e3b2ad3a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/ea1d82876704575f3a48125b30cb224e3b2ad3a3"}]},"branch":"refs/heads/master"},"1809f3b5d9b3256f45ed1e18f174c498479ae4a5":{"kind":"REWORK","_number":4,"created":"2019-06-03 14:50:55.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/40/662740/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/40/662740/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/4"}}},"commit":{"parents":[{"commit":"fde568564d43c78bf2828d3c981d5cc685fb3369","subject":"Add new pcs 0.10 compatible remote addition implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/fde568564d43c78bf2828d3c981d5cc685fb3369"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-06-03 12:17:55.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-06-03 14:50:46.000000000","tz":120},"subject":"WIP - Support for SBD (watchdog only)","message":"WIP - Support for SBD (watchdog only)\n\nThis commit allow the operators to configure SBD-based fencing.\n\nBecause enabling sbd requires a cluster restart the assumption\nis that sbd will be enabled at cluster creation time by setting\n\"pacemaker::corosync::enable_sbd\" to true.\n\nparameter_defaults:\n  ExtraConfig:\n    pacemaker::corosync::enable_sbd: true\n\nAn optional watchdog_timeout value (in seconds) can be also specified via:\n\n    pacemaker::corosync::sbd_watchdog_timeout: 20\n\nIf unset, the default value will be set to 10s.\n\nChange-Id: Iad1193c1df0792dcb7dce2af3575e19e58781ba0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/1809f3b5d9b3256f45ed1e18f174c498479ae4a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/1809f3b5d9b3256f45ed1e18f174c498479ae4a5"}]},"branch":"refs/heads/master"},"ecd9e2a1392cfbe18352c380976ff403d8916dd9":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-07-24 05:44:45.000000000","uploader":{"_account_id":30126,"name":"Luca Miccini","email":"lmiccini@redhat.com","username":"lmiccini2"},"ref":"refs/changes/40/662740/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/40/662740/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/5"}}},"commit":{"parents":[{"commit":"fde568564d43c78bf2828d3c981d5cc685fb3369","subject":"Add new pcs 0.10 compatible remote addition implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/fde568564d43c78bf2828d3c981d5cc685fb3369"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-06-03 12:17:55.000000000","tz":120},"committer":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-07-24 05:36:17.000000000","tz":120},"subject":"Support for SBD (watchdog only)","message":"Support for SBD (watchdog only)\n\nThis commit introduces minimal support for SBD-based fencing,\nlimited to watchdog devices (no shared storage).\n\nTargeted use case is operators wanting to employ watchdog as\nN-level stonith mechanism.\n\nBecause sbd requires a cluster restart the (initial) assumption\nis that sbd will be enabled at cluster creation time by setting\n\"pacemaker::corosync::enable_sbd\" to true.\n\nparameter_defaults:\n  ExtraConfig:\n    pacemaker::corosync::enable_sbd: true\n\nOptionally, the default watchdog_timeout value (10 seconds)\ncan be overridden like:\n\n    pacemaker::corosync::sbd_watchdog_timeout: 20\n\nThe 10s value is necessary (at least in virtual environments)\nto prevent fencing from happening during the deployment.\n\nChange-Id: Iad1193c1df0792dcb7dce2af3575e19e58781ba0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/ecd9e2a1392cfbe18352c380976ff403d8916dd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/ecd9e2a1392cfbe18352c380976ff403d8916dd9"}]},"branch":"refs/heads/master"},"a8653f23da958b9e688ac01b8175eff326086928":{"kind":"REWORK","_number":6,"created":"2019-08-02 08:33:36.000000000","uploader":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"ref":"refs/changes/40/662740/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/40/662740/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/6"}}},"commit":{"parents":[{"commit":"fde568564d43c78bf2828d3c981d5cc685fb3369","subject":"Add new pcs 0.10 compatible remote addition implementation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/fde568564d43c78bf2828d3c981d5cc685fb3369"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-06-03 12:17:55.000000000","tz":120},"committer":{"name":"Michele Baldessari","email":"michele@acksyn.org","date":"2019-08-02 08:33:01.000000000","tz":120},"subject":"Support for SBD (watchdog only)","message":"Support for SBD (watchdog only)\n\nThis commit introduces minimal support for SBD-based fencing,\nlimited to watchdog devices (no shared storage).\n\nTargeted use case is operators wanting to employ watchdog as\nN-level stonith mechanism.\n\nBecause sbd requires a cluster restart the (initial) assumption\nis that sbd will be enabled at cluster creation time by setting\n\"pacemaker::corosync::enable_sbd\" to true.\n\nparameter_defaults:\n  ExtraConfig:\n    pacemaker::corosync::enable_sbd: true\n\nOptionally, the default watchdog_timeout value (10 seconds)\ncan be overridden like:\n\n    pacemaker::corosync::sbd_watchdog_timeout: 20\n\nThe 10s value is necessary (at least in virtual environments)\nto prevent fencing from happening during the deployment.\n\nChange-Id: Iad1193c1df0792dcb7dce2af3575e19e58781ba0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/a8653f23da958b9e688ac01b8175eff326086928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/a8653f23da958b9e688ac01b8175eff326086928"}]},"branch":"refs/heads/master"},"5bf38785f6419653f771d16188e3fced5f24b70a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2019-08-02 16:22:03.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/40/662740/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/40/662740/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/7"}}},"commit":{"parents":[{"commit":"a0f4b94436647461380fb3a26605d3dfeeee463a","subject":"Pin rspec-puppet-facts if older ruby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/a0f4b94436647461380fb3a26605d3dfeeee463a"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-06-03 12:17:55.000000000","tz":120},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-02 16:22:03.000000000","tz":0},"subject":"Support for SBD (watchdog only)","message":"Support for SBD (watchdog only)\n\nThis commit introduces minimal support for SBD-based fencing,\nlimited to watchdog devices (no shared storage).\n\nTargeted use case is operators wanting to employ watchdog as\nN-level stonith mechanism.\n\nBecause sbd requires a cluster restart the (initial) assumption\nis that sbd will be enabled at cluster creation time by setting\n\"pacemaker::corosync::enable_sbd\" to true.\n\nparameter_defaults:\n  ExtraConfig:\n    pacemaker::corosync::enable_sbd: true\n\nOptionally, the default watchdog_timeout value (10 seconds)\ncan be overridden like:\n\n    pacemaker::corosync::sbd_watchdog_timeout: 20\n\nThe 10s value is necessary (at least in virtual environments)\nto prevent fencing from happening during the deployment.\n\nChange-Id: Iad1193c1df0792dcb7dce2af3575e19e58781ba0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/5bf38785f6419653f771d16188e3fced5f24b70a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/5bf38785f6419653f771d16188e3fced5f24b70a"}]},"branch":"refs/heads/master"},"519c28b163cef8ef21e83d7c177c0936de55e183":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2019-08-02 16:40:48.000000000","uploader":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"ref":"refs/changes/40/662740/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/40/662740/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/40/662740/8"}}},"commit":{"parents":[{"commit":"ffbed58789fbf290817d68ca24b931098661bac0","subject":"Pin rspec-puppet-facts if older ruby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/ffbed58789fbf290817d68ca24b931098661bac0"}]}],"author":{"name":"Luca Miccini","email":"lmiccini@redhat.com","date":"2019-06-03 12:17:55.000000000","tz":120},"committer":{"name":"Alex Schultz","email":"aschultz@redhat.com","date":"2019-08-02 16:40:48.000000000","tz":0},"subject":"Support for SBD (watchdog only)","message":"Support for SBD (watchdog only)\n\nThis commit introduces minimal support for SBD-based fencing,\nlimited to watchdog devices (no shared storage).\n\nTargeted use case is operators wanting to employ watchdog as\nN-level stonith mechanism.\n\nBecause sbd requires a cluster restart the (initial) assumption\nis that sbd will be enabled at cluster creation time by setting\n\"pacemaker::corosync::enable_sbd\" to true.\n\nparameter_defaults:\n  ExtraConfig:\n    pacemaker::corosync::enable_sbd: true\n\nOptionally, the default watchdog_timeout value (10 seconds)\ncan be overridden like:\n\n    pacemaker::corosync::sbd_watchdog_timeout: 20\n\nThe 10s value is necessary (at least in virtual environments)\nto prevent fencing from happening during the deployment.\n\nChange-Id: Iad1193c1df0792dcb7dce2af3575e19e58781ba0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/519c28b163cef8ef21e83d7c177c0936de55e183"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/519c28b163cef8ef21e83d7c177c0936de55e183"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
