)]}'
{"id":"openstack%2Fpuppet-pacemaker~671011","triplet_id":"openstack%2Fpuppet-pacemaker~master~If5baef83c739c7ccdbe346c75401f232ab092d20","project":"openstack/puppet-pacemaker","branch":"master","topic":"pcs-010-node-add","hashtags":[],"change_id":"If5baef83c739c7ccdbe346c75401f232ab092d20","subject":"Scale up logic with extra IP addresses for pcs 0.10","status":"MERGED","created":"2019-07-16 11:29:14.000000000","updated":"2019-07-19 13:03:52.000000000","submitted":"2019-07-19 12:44:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671011-1563540297337-0a1ee91c","meta_rev_id":"7608bc57d416bc455034c8e05cfad2b987471d29","_number":671011,"virtual_id_number":671011,"owner":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"date":"2019-07-18 15:12:56.000000000","_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-19 12:44:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-07-17 16:18:13.000000000","_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-19 12:27:22.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":2,"date":"2019-07-18 15:51:47.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":1,"date":"2019-07-17 13:43:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-07-19 12:27:22.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-07-18 13:20:39.000000000","_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"value":0,"date":"2019-07-18 13:20:41.000000000","_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},{"value":0,"date":"2019-07-18 13:20:49.000000000","_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"date":"2019-07-17 13:24:09.000000000","_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"value":0,"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"value":0,"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"value":0,"date":"2019-07-18 15:11:54.000000000","_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-17 13:24:09.000000000","updated_by":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"reviewer":{"_account_id":20778,"name":"Damien Ciabrini","email":"dciabrin@redhat.com","username":"dciabrin"},"state":"REVIEWER"},{"updated":"2019-07-17 13:43:05.000000000","updated_by":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"reviewer":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"state":"REVIEWER"},{"updated":"2019-07-17 16:18:13.000000000","updated_by":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"reviewer":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-07-18 13:20:39.000000000","updated_by":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"reviewer":{"_account_id":20775,"name":"Carlos Camacho","email":"ccamacho@redhat.com","username":"ccamacho"},"state":"REVIEWER"},{"updated":"2019-07-18 13:20:41.000000000","updated_by":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"reviewer":{"_account_id":26343,"name":"Jose Luis Franco","email":"jfrancoa@redhat.com","username":"jfrancoa"},"state":"REVIEWER"},{"updated":"2019-07-18 13:20:49.000000000","updated_by":{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},"reviewer":{"_account_id":8297,"name":"Sofer Athlan-Guyot","email":"sathlang@redhat.com","username":"chem"},"state":"REVIEWER"},{"updated":"2019-07-18 15:11:54.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-18 15:51:47.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-07-19 12:27:22.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-19 12:44:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be6a0e218703a79eb9513611e44c302888a1efc3","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2019-07-16 11:29:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f685757e25e1e541181a790c07b1d54421a64be8","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2019-07-16 11:29:29.000000000","message":"Patch Set 1: Workflow-1\n\nWIP","accounts_in_message":[],"_revision_number":1},{"id":"ca65c3548c9483e0b0f737967972b5c36a726b32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 11:51:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/11/671011/1/check/puppet-openstack-lint-ubuntu-bionic/186a53b/ : SUCCESS in 5m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/11/671011/1/check/puppet-openstack-syntax-5-ubuntu-bionic/3ecdd7d/ : SUCCESS in 5m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/11/671011/1/check/puppet-openstack-syntax-6-ubuntu-bionic/2aab2a1/ : SUCCESS in 5m 21s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/11/671011/1/check/puppet-openstack-unit-5.5-centos-7/8454801/ : SUCCESS in 12m 17s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/11/671011/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/b5fccc6/ : SUCCESS in 8m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/11/671011/1/check/puppet-openstack-unit-latest-ubuntu-bionic/2fb26b9/ : SUCCESS in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce26dbf393e5855ccb983804c81be2e47aa23d5d","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-16 14:11:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/11/671011/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/9f83047/ : SUCCESS in 2h 32m 24s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/11/671011/1/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/06e9e11/ : SUCCESS in 2h 40m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6c7d6616697623b4f78c1075235c646c80ab1fcf","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2019-07-16 14:59:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d1e89eee09f017b4c1e226f296e6cd384d12a09d","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2019-07-16 14:59:29.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"32b90ec8518a1811e8cee51e3c70c5d2a2952e09","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-16 15:32:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/11/671011/2/check/puppet-openstack-lint-ubuntu-bionic/d29602d/ : SUCCESS in 5m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/11/671011/2/check/puppet-openstack-syntax-5-ubuntu-bionic/af38ae4/ : SUCCESS in 5m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/11/671011/2/check/puppet-openstack-syntax-6-ubuntu-bionic/5b05d6e/ : SUCCESS in 6m 59s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/11/671011/2/check/puppet-openstack-unit-5.5-centos-7/bf98779/ : SUCCESS in 14m 47s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/11/671011/2/check/puppet-openstack-unit-6.4-ubuntu-bionic/52388e1/ : SUCCESS in 11m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/11/671011/2/check/puppet-openstack-unit-latest-ubuntu-bionic/763ea55/ : SUCCESS in 8m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e2d8b05bb64839df5d7dc5369b2d271bec0479ae","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-16 16:24:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 http://logs.rdoproject.org/11/671011/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001/af4e352/ : FAILURE in 1h 06m 52s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/11/671011/2/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/04a0b0c/ : FAILURE in 1h 07m 28s","accounts_in_message":[],"_revision_number":2},{"id":"492fc15f19b607eb735a61a552220ae6c94b3290","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2019-07-17 13:23:39.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6d054c3d070ec3aa801a3ae76083b98841657085","author":{"_account_id":20172,"name":"Michele Baldessari","email":"michele@acksyn.org","username":"michele"},"date":"2019-07-17 13:43:05.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cf3d73c4aff6dc7c5aa2767f9ebef78b28f0fb93","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-17 13:53:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/11/671011/3/check/puppet-openstack-lint-ubuntu-bionic/338ad24/ : SUCCESS in 5m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/11/671011/3/check/puppet-openstack-syntax-5-ubuntu-bionic/c121892/ : SUCCESS in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/11/671011/3/check/puppet-openstack-syntax-6-ubuntu-bionic/99bc97a/ : SUCCESS in 5m 50s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/11/671011/3/check/puppet-openstack-unit-5.5-centos-7/bf3fe11/ : SUCCESS in 9m 28s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/11/671011/3/check/puppet-openstack-unit-6.4-ubuntu-bionic/2a52c6e/ : SUCCESS in 14m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/11/671011/3/check/puppet-openstack-unit-latest-ubuntu-bionic/640687c/ : SUCCESS in 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0cd06e54da56bc61593f61c10f6c348bf546382a","author":{"_account_id":23181,"name":"RDO Third Party CI","email":"dmsimard+rdothirdparty@redhat.com","username":"rdothirdparty","tags":["SERVICE_USER"]},"date":"2019-07-17 16:18:13.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset001 finger://ze02.softwarefactory-project.io:7979/2f0b05cf1d7b4dde8d4cd5a50539ab6f : RETRY_LIMIT in 6s\n- tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035 http://logs.rdoproject.org/11/671011/3/openstack-check/tripleo-ci-centos-7-ovb-3ctlr_1comp-featureset035/06dc123/ : SUCCESS in 2h 50m 47s","accounts_in_message":[],"_revision_number":3},{"id":"d016e674da90322e3cf9483c31bf8094df3a7f2d","author":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"date":"2019-07-18 15:12:56.000000000","message":"Patch Set 3:\n\nI tested this manually within a Rocky-\u003eStein upgrade FWIW.","accounts_in_message":[],"_revision_number":3},{"id":"6866708dae4a91fbf0afaf1969133ed3417cd49f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-07-18 15:51:47.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"84fb3b61181dc71b0baee28757cbcd9f93d0a565","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-19 12:27:22.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cadfa2fc671aa858ccd55cb1ec9bd2cb3ce73f94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:27:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1d88d344e0f945a5f9a96e5402c25278a4dcabff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:44:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7b59204617e21f24b70d830893306f902b5009d2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 12:44:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/11/671011/3/gate/puppet-openstack-lint-ubuntu-bionic/edfbddf/ : SUCCESS in 5m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/11/671011/3/gate/puppet-openstack-syntax-5-ubuntu-bionic/81a1097/ : SUCCESS in 4m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/11/671011/3/gate/puppet-openstack-syntax-6-ubuntu-bionic/8bd873d/ : SUCCESS in 4m 07s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/11/671011/3/gate/puppet-openstack-unit-5.5-centos-7/04f6053/ : SUCCESS in 11m 03s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/11/671011/3/gate/puppet-openstack-unit-6.4-ubuntu-bionic/5b26bf1/ : SUCCESS in 8m 56s","accounts_in_message":[],"_revision_number":3},{"id":"3fe851e6229294a97f5e113e1317692d2d05468c","author":{"_account_id":11090,"name":"Sergii Golovatiuk","email":"sgolovat@redhat.com","username":"holser"},"date":"2019-07-19 13:03:52.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/0.7.x as commit d7e869fd30eba8309e35343f9d7e98e527d81c77","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d0b394d8f64d779c4866386f11f2c4c48f299ce5","revisions":{"96e10f4154167d9da8c0d73ffc0c8f3e3bbcbaa1":{"kind":"REWORK","_number":1,"created":"2019-07-16 11:29:14.000000000","uploader":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"ref":"refs/changes/11/671011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/11/671011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/11/671011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/11/671011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/11/671011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/11/671011/1"}}},"commit":{"parents":[{"commit":"7ddbe0860e7544d1d66fb727a4465045f224c741","subject":"Force to use markdown to prevent pypi issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/7ddbe0860e7544d1d66fb727a4465045f224c741"}]}],"author":{"name":"Jiri Stransky","email":"jistr@redhat.com","date":"2019-07-16 11:27:56.000000000","tz":120},"committer":{"name":"Jiri Stransky","email":"jistr@redhat.com","date":"2019-07-16 11:29:13.000000000","tz":120},"subject":"Scale up logic with extra IP addresses for pcs 0.10","message":"Scale up logic with extra IP addresses for pcs 0.10\n\nChange-Id: If5baef83c739c7ccdbe346c75401f232ab092d20\nRelated-Bug: #1831690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/96e10f4154167d9da8c0d73ffc0c8f3e3bbcbaa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/96e10f4154167d9da8c0d73ffc0c8f3e3bbcbaa1"}]},"branch":"refs/heads/master"},"63aa21feeafecd8840de15472125a67b939f0f3d":{"kind":"REWORK","_number":2,"created":"2019-07-16 14:59:03.000000000","uploader":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"ref":"refs/changes/11/671011/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/11/671011/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/11/671011/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/11/671011/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/11/671011/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/11/671011/2"}}},"commit":{"parents":[{"commit":"7ddbe0860e7544d1d66fb727a4465045f224c741","subject":"Force to use markdown to prevent pypi issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/7ddbe0860e7544d1d66fb727a4465045f224c741"}]}],"author":{"name":"Jiri Stransky","email":"jistr@redhat.com","date":"2019-07-16 11:27:56.000000000","tz":120},"committer":{"name":"Jiri Stransky","email":"jistr@redhat.com","date":"2019-07-16 14:32:05.000000000","tz":120},"subject":"Scale up logic with extra IP addresses for pcs 0.10","message":"Scale up logic with extra IP addresses for pcs 0.10\n\nChange-Id: If5baef83c739c7ccdbe346c75401f232ab092d20\nRelated-Bug: #1831690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/63aa21feeafecd8840de15472125a67b939f0f3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/63aa21feeafecd8840de15472125a67b939f0f3d"}]},"branch":"refs/heads/master"},"d0b394d8f64d779c4866386f11f2c4c48f299ce5":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-07-17 13:23:39.000000000","uploader":{"_account_id":8042,"name":"Jiří Stránský","email":"jistr@redhat.com","username":"jistr"},"ref":"refs/changes/11/671011/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-pacemaker","ref":"refs/changes/11/671011/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/11/671011/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/11/671011/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-pacemaker refs/changes/11/671011/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-pacemaker refs/changes/11/671011/3"}}},"commit":{"parents":[{"commit":"7ddbe0860e7544d1d66fb727a4465045f224c741","subject":"Force to use markdown to prevent pypi issue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/7ddbe0860e7544d1d66fb727a4465045f224c741"}]}],"author":{"name":"Jiri Stransky","email":"jistr@redhat.com","date":"2019-07-16 11:27:56.000000000","tz":120},"committer":{"name":"Jiri Stransky","email":"jistr@redhat.com","date":"2019-07-17 13:18:41.000000000","tz":120},"subject":"Scale up logic with extra IP addresses for pcs 0.10","message":"Scale up logic with extra IP addresses for pcs 0.10\n\nNowadays we can provide IP addresses when adding a host into a\npacemaker cluster. The pcmk_cluster_setup function for initial setup\nof the cluster has this already implemented, but it was still missing\nfrom the pcmk_nodes_added function and the scale-up logic. Support for\nscaling up with pcs 0.10 and the explicitly provided IP addresses is\nbeing added in this commit.\n\nChange-Id: If5baef83c739c7ccdbe346c75401f232ab092d20\nCloses-Bug: #1835217\nRelated-Bug: #1831690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/d0b394d8f64d779c4866386f11f2c4c48f299ce5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-pacemaker/commit/d0b394d8f64d779c4866386f11f2c4c48f299ce5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
