)]}'
{"id":"openstack%2Fpuppet-panko~654215","triplet_id":"openstack%2Fpuppet-panko~master~I7973b26793094052f8d9eb0d6eaf9722ef2df9d9","project":"openstack/puppet-panko","branch":"master","topic":"move-to-opendev","hashtags":[],"change_id":"I7973b26793094052f8d9eb0d6eaf9722ef2df9d9","subject":"Move to opendev","status":"MERGED","created":"2019-04-21 06:47:01.000000000","updated":"2019-04-22 18:18:53.000000000","submitted":"2019-04-22 18:18:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"654215-1555957133486-a663f06a","meta_rev_id":"ea3dce21924693209bf96efac1830701dea707fa","_number":654215,"virtual_id_number":654215,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-04-21 08:56:31.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-04-22 18:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-04-22 13:22:50.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-04-22 17:05:53.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2019-04-22 17:05:53.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-22 13:22:50.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-22 17:05:53.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-04-22 18:18:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab2b542c27e2761c3919ee85aa53bbc2ca370bb6","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 06:47:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7335cbc6db8899fc0d5bd1008b375b569f50432","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 08:12:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-lint-ubuntu-bionic/c538197/ : SUCCESS in 4m 27s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-syntax-4-ubuntu-bionic/b026655/ : SUCCESS in 4m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-syntax-5-ubuntu-bionic/af53c56/ : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-syntax-6-ubuntu-bionic/631e9e7/ : SUCCESS in 4m 43s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/15/654215/1/check/puppet-openstack-unit-4.8-centos-7/e029ce9/ : SUCCESS in 16m 52s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/15/654215/1/check/puppet-openstack-unit-5.5-centos-7/3ed8d99/ : SUCCESS in 15m 47s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/314236b/ : SUCCESS in 8m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-unit-latest-ubuntu-bionic/cc49fea/ : FAILURE in 8m 02s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/15/654215/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/b7e1a02/ : FAILURE in 34m 47s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/15/654215/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/e62b549/ : FAILURE in 1h 02m 57s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/15/654215/1/check/puppet-openstack-beaker-centos-7/f06512b/ : SUCCESS in 15m 20s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-beaker-ubuntu-bionic/7872b24/ : FAILURE in 10m 39s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/15/654215/1/check/build-openstack-releasenotes/46ec3d7/html/ : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b5acac5e8ae5574c469412a11576bfa4b6a5879a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 08:15:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"06982d29654800c5e974812e21acd522467368c4","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-21 08:56:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"db60f105c7da0a9e324ecf3890e618681a789090","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-21 09:13:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-lint-ubuntu-bionic/110e6d0/ : SUCCESS in 4m 07s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-syntax-4-ubuntu-bionic/fd4e19e/ : SUCCESS in 3m 50s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-syntax-5-ubuntu-bionic/636bd90/ : SUCCESS in 6m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-syntax-6-ubuntu-bionic/ec65014/ : SUCCESS in 3m 44s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/15/654215/1/check/puppet-openstack-unit-4.8-centos-7/e9f7ba2/ : SUCCESS in 16m 06s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/15/654215/1/check/puppet-openstack-unit-5.5-centos-7/c7c354e/ : SUCCESS in 15m 31s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/2e502f4/ : SUCCESS in 9m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-unit-latest-ubuntu-bionic/7ae8dae/ : FAILURE in 7m 33s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/15/654215/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/b1b8b32/ : FAILURE in 30m 19s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/15/654215/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/c548e4d/ : SUCCESS in 53m 11s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/15/654215/1/check/puppet-openstack-beaker-centos-7/39ec283/ : SUCCESS in 13m 54s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/15/654215/1/check/puppet-openstack-beaker-ubuntu-bionic/2300a34/ : FAILURE in 12m 17s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/15/654215/1/check/build-openstack-releasenotes/0aa949a/html/ : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":1},{"id":"666b36cb82e4585a71abb8c08d2925b59369aa0e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-22 13:22:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"44364c2d92718687aafe6c2fb2b97faa993d99c4","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-04-22 17:05:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d6bd4b63c62895950dbee599b686f538baa979dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 17:06:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac03c72434d10ec2b4d373223596d399be155a6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:18:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a8df52351b7b4cb494bc38c197c22c9499ab1693","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:18:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/15/654215/1/gate/puppet-openstack-lint-ubuntu-bionic/a739153/ : SUCCESS in 3m 56s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/15/654215/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/0b401c0/ : SUCCESS in 4m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/15/654215/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/3cdc283/ : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/15/654215/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/b0c9792/ : SUCCESS in 4m 50s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/15/654215/1/gate/puppet-openstack-unit-4.8-centos-7/68c8b0a/ : SUCCESS in 10m 57s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/15/654215/1/gate/puppet-openstack-unit-5.5-centos-7/2d2157f/ : SUCCESS in 10m 23s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/15/654215/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/a443b4d/ : SUCCESS in 6m 59s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/15/654215/1/gate/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/1a1e9ac/ : FAILURE in 34m 53s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/15/654215/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/ec6958d/ : SUCCESS in 1h 01m 29s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/15/654215/1/gate/puppet-openstack-beaker-centos-7/cd9cad2/ : SUCCESS in 17m 14s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/15/654215/1/gate/puppet-openstack-beaker-ubuntu-bionic/fe54c3d/ : FAILURE in 10m 29s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/15/654215/1/gate/build-openstack-releasenotes/29bae5d/html/ : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca13338ea23875e03bf81e4cec4cbef8c39254fa","revisions":{"ca13338ea23875e03bf81e4cec4cbef8c39254fa":{"kind":"REWORK","_number":1,"created":"2019-04-21 06:47:01.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/15/654215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/15/654215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/15/654215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/15/654215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/15/654215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/15/654215/1"}}},"commit":{"parents":[{"commit":"4c07b75b12efa20504e5953a0160fe27f5f9d79c","subject":"Fix Repository","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/4c07b75b12efa20504e5953a0160fe27f5f9d79c"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:49.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-21 06:44:49.000000000","tz":480},"subject":"Move to opendev","message":"Move to opendev\n\nChange-Id: I7973b26793094052f8d9eb0d6eaf9722ef2df9d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ca13338ea23875e03bf81e4cec4cbef8c39254fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ca13338ea23875e03bf81e4cec4cbef8c39254fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
