)]}'
{"id":"openstack%2Fpuppet-panko~656122","triplet_id":"openstack%2Fpuppet-panko~master~I684b2cd4973adc7d4ece83e620fd053e191bb19f","project":"openstack/puppet-panko","branch":"master","topic":"bug/1826692","hashtags":[],"change_id":"I684b2cd4973adc7d4ece83e620fd053e191bb19f","subject":"Deprecate idle_timeout option","status":"MERGED","created":"2019-04-28 07:00:56.000000000","updated":"2019-04-29 08:28:44.000000000","submitted":"2019-04-29 08:28:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"656122-1556526524988-7d7fc31e","meta_rev_id":"28eb9aa7a54cac55294ff37a918516236ffdfb5f","_number":656122,"virtual_id_number":656122,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2019-04-29 06:05:47.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2019-04-29 08:28:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-04-28 08:43:11.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-04-28 17:13:31.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-04-28 17:13:31.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-28 08:43:11.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-04-28 17:13:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-04-29 08:28:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6014447546d286d797fb35ee067c191325440758","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-28 07:00:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ee793e8b6ce7afed4b8dd6051c2738c180adf00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-28 08:07:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-lint-ubuntu-bionic/bd2d3ea/ : SUCCESS in 5m 05s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-syntax-4-ubuntu-bionic/44126b3/ : SUCCESS in 5m 11s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-syntax-5-ubuntu-bionic/7168c12/ : SUCCESS in 5m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-syntax-6-ubuntu-bionic/5fb005e/ : SUCCESS in 4m 52s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/22/656122/1/check/puppet-openstack-unit-4.8-centos-7/75fc659/ : SUCCESS in 13m 00s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/22/656122/1/check/puppet-openstack-unit-5.5-centos-7/e9349b7/ : SUCCESS in 9m 35s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/f2af3f1/ : SUCCESS in 8m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-unit-latest-ubuntu-bionic/d185181/ : SUCCESS in 11m 29s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/22/656122/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/20515aa/ : FAILURE in 30m 48s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/22/656122/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/3482056/ : SUCCESS in 1h 00m 02s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/22/656122/1/check/puppet-openstack-beaker-centos-7/7ff964f/ : SUCCESS in 14m 19s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-beaker-ubuntu-bionic/523c4a7/ : FAILURE in 11m 14s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/22/656122/1/check/build-openstack-releasenotes/ae62d79/html/ : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bd111efe167a7d1c0c31ecd9f51f454848e47838","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-04-28 08:43:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a2a2810c0aaa2bb6efc36149392ba899b75e7223","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-04-28 17:13:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eaadf27fc94fd00c546a86b078dae8b864c64f82","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-04-29 06:05:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"86b09be3193fb6ffd66e67e7f204ddb07dd1d2b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 07:13:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-lint-ubuntu-bionic/faed997/ : SUCCESS in 6m 45s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-syntax-4-ubuntu-bionic/8db8362/ : SUCCESS in 6m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-syntax-5-ubuntu-bionic/eef2f13/ : SUCCESS in 6m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-syntax-6-ubuntu-bionic/be53495/ : SUCCESS in 6m 35s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/22/656122/1/check/puppet-openstack-unit-4.8-centos-7/6b71b6b/ : SUCCESS in 9m 48s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/22/656122/1/check/puppet-openstack-unit-5.5-centos-7/57b2bc3/ : SUCCESS in 9m 58s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-unit-6.0-ubuntu-bionic/fbb37e8/ : SUCCESS in 8m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-unit-latest-ubuntu-bionic/05bc609/ : SUCCESS in 7m 22s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/22/656122/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/0bf0dd8/ : FAILURE in 30m 10s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/22/656122/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/c82a4fa/ : SUCCESS in 54m 31s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/22/656122/1/check/puppet-openstack-beaker-centos-7/a18d28b/ : SUCCESS in 17m 14s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/22/656122/1/check/puppet-openstack-beaker-ubuntu-bionic/5fd0e59/ : FAILURE in 7m 09s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/22/656122/1/check/build-openstack-releasenotes/24652eb/html/ : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"efc92994ffd9e76c11d77f1a9cc16ce5f7a2f971","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 07:13:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac65c106cdee9d2bfa0ea00869c9882aa159bbd1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 08:28:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"41fc13f6a24b8618a3e14d58eff2c791a3316a29","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-29 08:28:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/22/656122/1/gate/puppet-openstack-lint-ubuntu-bionic/a026abf/ : SUCCESS in 4m 35s\n- puppet-openstack-syntax-4-ubuntu-bionic http://logs.openstack.org/22/656122/1/gate/puppet-openstack-syntax-4-ubuntu-bionic/5b05ae4/ : SUCCESS in 3m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/22/656122/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/df5a810/ : SUCCESS in 3m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/22/656122/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/10bf50d/ : SUCCESS in 3m 26s\n- puppet-openstack-unit-4.8-centos-7 http://logs.openstack.org/22/656122/1/gate/puppet-openstack-unit-4.8-centos-7/3a0769d/ : SUCCESS in 10m 10s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/22/656122/1/gate/puppet-openstack-unit-5.5-centos-7/0a05805/ : SUCCESS in 9m 56s\n- puppet-openstack-unit-6.0-ubuntu-bionic http://logs.openstack.org/22/656122/1/gate/puppet-openstack-unit-6.0-ubuntu-bionic/edd0b0c/ : SUCCESS in 5m 53s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/22/656122/1/gate/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/ef95a5d/ : FAILURE in 37m 12s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/22/656122/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/cad7a1a/ : SUCCESS in 1h 02m 55s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/22/656122/1/gate/puppet-openstack-beaker-centos-7/763c534/ : SUCCESS in 15m 02s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/22/656122/1/gate/puppet-openstack-beaker-ubuntu-bionic/0f19a71/ : FAILURE in 6m 25s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/22/656122/1/gate/build-openstack-releasenotes/c70a119/html/ : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ad0b8af0f40fc075cb57d4a0e87e46af27a4ba31","revisions":{"ad0b8af0f40fc075cb57d4a0e87e46af27a4ba31":{"kind":"REWORK","_number":1,"created":"2019-04-28 07:00:56.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/22/656122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/22/656122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/22/656122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/22/656122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/22/656122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/22/656122/1"}}},"commit":{"parents":[{"commit":"ca13338ea23875e03bf81e4cec4cbef8c39254fa","subject":"Move to opendev","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ca13338ea23875e03bf81e4cec4cbef8c39254fa"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 07:00:12.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-04-28 07:00:12.000000000","tz":480},"subject":"Deprecate idle_timeout option","message":"Deprecate idle_timeout option\n\nThe idle_timeout parameter is deprecated, use connection_recycle_time\ninstead[1].\n\n[1]https://review.opendev.org/#/c/334182/\n\nChange-Id: I684b2cd4973adc7d4ece83e620fd053e191bb19f\nDepends-On: https://review.opendev.org/656106/\nCloses-Bug: #1826692\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ad0b8af0f40fc075cb57d4a0e87e46af27a4ba31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ad0b8af0f40fc075cb57d4a0e87e46af27a4ba31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
