)]}'
{"id":"openstack%2Fpuppet-panko~663203","triplet_id":"openstack%2Fpuppet-panko~master~Idb691484e7e4f3d36adf7629c351952e4cab2070","project":"openstack/puppet-panko","branch":"master","topic":"prep-release-train-1","hashtags":[],"change_id":"Idb691484e7e4f3d36adf7629c351952e4cab2070","subject":"Prepare Train M2","status":"MERGED","created":"2019-06-05 01:42:59.000000000","updated":"2019-06-10 13:27:02.000000000","submitted":"2019-06-10 13:27:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"663203-1560173222388-e5d355c6","meta_rev_id":"19382cce4715080f9e1a10a7ee27e70404cf3497","_number":663203,"virtual_id_number":663203,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2019-06-10 13:27:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-06-10 13:22:08.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-06-10 13:22:08.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-10 13:22:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-06-10 13:27:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a9d6fd43b1dfe2840ba54d577b8f78bc0a70785","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-05 01:42:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3bd4931942d11f25c3170d618ba91cf27b22bbc7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 02:09:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/03/663203/1/check/build-openstack-puppet-tarball/ce8c4b4/ : SUCCESS in 2m 55s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/03/663203/1/check/puppet-openstack-lint-ubuntu-bionic/783182c/ : SUCCESS in 3m 58s","accounts_in_message":[],"_revision_number":1},{"id":"580a2578c11f17d757a3b8ce1f379789cd964bbd","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-06-10 02:09:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ea1be701ae409cdc27a718b67de83691b5e68167","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 02:45:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/03/663203/2/check/build-openstack-puppet-tarball/4d352c7/ : SUCCESS in 2m 13s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/03/663203/2/check/puppet-openstack-lint-ubuntu-bionic/6e7e2bf/ : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":2},{"id":"b25be7d1964e732a8b541a72f9a0b3371ae88208","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-06-10 13:22:08.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8b7ff2abbab77b77a7c6f89f71f2379161ff377d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:22:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2af3ce7d3c818c207762b620b961fb5234a74089","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:27:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"455e943aa34827d7fa991ae3fae809e2ada4b5c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 13:27:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball http://logs.openstack.org/03/663203/2/gate/build-openstack-puppet-tarball/08a6f2a/ : SUCCESS in 2m 02s\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/03/663203/2/gate/puppet-openstack-lint-ubuntu-bionic/26e6965/ : SUCCESS in 4m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5f69567dc7327702fc037104a65051c20cd478da","revisions":{"cc8a0b431be07e14db4c42a82a48d4f800796068":{"kind":"REWORK","_number":1,"created":"2019-06-05 01:42:59.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/03/663203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/03/663203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/03/663203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/03/663203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/03/663203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/03/663203/1"}}},"commit":{"parents":[{"commit":"ad0b8af0f40fc075cb57d4a0e87e46af27a4ba31","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ad0b8af0f40fc075cb57d4a0e87e46af27a4ba31"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:22.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:22.000000000","tz":480},"subject":"Prepare Train M1","message":"Prepare Train M1\n\nUpdate the version metadata for Train milestone 1\n\nChange-Id: Idb691484e7e4f3d36adf7629c351952e4cab2070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/cc8a0b431be07e14db4c42a82a48d4f800796068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/cc8a0b431be07e14db4c42a82a48d4f800796068"}]},"branch":"refs/heads/master"},"5f69567dc7327702fc037104a65051c20cd478da":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-10 02:09:18.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/03/663203/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/03/663203/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/03/663203/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/03/663203/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/03/663203/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/03/663203/2"}}},"commit":{"parents":[{"commit":"ad0b8af0f40fc075cb57d4a0e87e46af27a4ba31","subject":"Deprecate idle_timeout option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ad0b8af0f40fc075cb57d4a0e87e46af27a4ba31"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-05 01:40:22.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-06-10 02:05:11.000000000","tz":480},"subject":"Prepare Train M2","message":"Prepare Train M2\n\nUpdate the version metadata for Train milestone 2\n\nChange-Id: Idb691484e7e4f3d36adf7629c351952e4cab2070\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/5f69567dc7327702fc037104a65051c20cd478da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/5f69567dc7327702fc037104a65051c20cd478da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
