)]}'
{"id":"openstack%2Fpuppet-panko~669040","triplet_id":"openstack%2Fpuppet-panko~master~Ifc011b5a261b7dfd13edf606c8f3cc711ca53489","project":"openstack/puppet-panko","branch":"master","topic":"bug/1835161","hashtags":[],"change_id":"Ifc011b5a261b7dfd13edf606c8f3cc711ca53489","subject":"Manage oslo_middleware/max_request_body_size","status":"MERGED","created":"2019-07-04 03:44:21.000000000","updated":"2019-07-05 01:55:00.000000000","submitted":"2019-07-05 01:55:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669040-1562291700961-eaa21e4c","meta_rev_id":"1c1846b6fc530efabb59ec05a5cfd9c9c896f7fd","_number":669040,"virtual_id_number":669040,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-07-05 01:55:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-07-04 14:02:59.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-07-05 00:51:23.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-04 14:02:59.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-07-05 01:55:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e7b4fbdf8c95d009238ffe9b406dba279c89a65","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-04 03:44:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aec68506d9f585ddae200e2de71a06a22d243824","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-04 04:53:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/40/669040/1/check/puppet-openstack-lint-ubuntu-bionic/bfe7c0e/ : SUCCESS in 6m 09s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/40/669040/1/check/puppet-openstack-syntax-5-ubuntu-bionic/a753cc5/ : SUCCESS in 4m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/40/669040/1/check/puppet-openstack-syntax-6-ubuntu-bionic/794e94e/ : SUCCESS in 6m 09s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/40/669040/1/check/puppet-openstack-unit-5.5-centos-7/63866c9/ : SUCCESS in 12m 22s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/40/669040/1/check/puppet-openstack-unit-6.4-ubuntu-bionic/d7bd4a7/ : SUCCESS in 12m 23s\n- puppet-openstack-unit-latest-ubuntu-bionic http://logs.openstack.org/40/669040/1/check/puppet-openstack-unit-latest-ubuntu-bionic/9b7f0d8/ : SUCCESS in 9m 17s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/40/669040/1/check/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/2d6b5d7/ : FAILURE in 6m 31s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/40/669040/1/check/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/119cf12/ : SUCCESS in 56m 11s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/40/669040/1/check/puppet-openstack-beaker-centos-7/750af62/ : SUCCESS in 16m 18s\n- puppet-openstack-beaker-ubuntu-bionic http://logs.openstack.org/40/669040/1/check/puppet-openstack-beaker-ubuntu-bionic/81de5fc/ : FAILURE in 9m 05s (non-voting)\n- build-openstack-releasenotes http://logs.openstack.org/40/669040/1/check/build-openstack-releasenotes/0dab65b/html/ : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e63ea61e6fe2766203a02b097444a7bdc6e21986","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-07-04 14:02:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"71ca1e216633e343049b06b5a736861187368173","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-07-05 00:51:23.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e16af9ebaf13f2b8f0b3e5a461b6d5d5616452a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 00:51:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cff026afe3305ad493399e2bf3e44b1de9b9b747","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 01:55:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8d093f438eb9daef16f5047fd92094b03900e9b6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-05 01:55:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic http://logs.openstack.org/40/669040/1/gate/puppet-openstack-lint-ubuntu-bionic/f30c456/ : SUCCESS in 6m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic http://logs.openstack.org/40/669040/1/gate/puppet-openstack-syntax-5-ubuntu-bionic/805702c/ : SUCCESS in 6m 26s\n- puppet-openstack-syntax-6-ubuntu-bionic http://logs.openstack.org/40/669040/1/gate/puppet-openstack-syntax-6-ubuntu-bionic/01e4266/ : SUCCESS in 5m 24s\n- puppet-openstack-unit-5.5-centos-7 http://logs.openstack.org/40/669040/1/gate/puppet-openstack-unit-5.5-centos-7/3b2607d/ : SUCCESS in 10m 59s\n- puppet-openstack-unit-6.4-ubuntu-bionic http://logs.openstack.org/40/669040/1/gate/puppet-openstack-unit-6.4-ubuntu-bionic/5a4d783/ : SUCCESS in 8m 21s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic http://logs.openstack.org/40/669040/1/gate/puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic/4be4673/ : FAILURE in 33m 22s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous http://logs.openstack.org/40/669040/1/gate/puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous/64e251c/ : SUCCESS in 55m 21s\n- puppet-openstack-beaker-centos-7 http://logs.openstack.org/40/669040/1/gate/puppet-openstack-beaker-centos-7/fa8b0ab/ : SUCCESS in 16m 23s\n- build-openstack-releasenotes http://logs.openstack.org/40/669040/1/gate/build-openstack-releasenotes/045eb97/html/ : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dfdb5f7d5370f24d77b25d5dd81040700aa4aeb2","revisions":{"dfdb5f7d5370f24d77b25d5dd81040700aa4aeb2":{"kind":"REWORK","_number":1,"created":"2019-07-04 03:44:21.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/40/669040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/40/669040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/669040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/669040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/669040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/40/669040/1"}}},"commit":{"parents":[{"commit":"5f69567dc7327702fc037104a65051c20cd478da","subject":"Prepare Train M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/5f69567dc7327702fc037104a65051c20cd478da"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-04 03:34:44.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-07-04 03:37:14.000000000","tz":480},"subject":"Manage oslo_middleware/max_request_body_size","message":"Manage oslo_middleware/max_request_body_size\n\nSo that we can increase it from the default 114688\n\nUseful in case for example the OS-Federation mapping is too large.\n\nIf this limit is breached panko will return a 413 Entity Too Large\nand not log anything to panko.log.\n\nChange-Id: Ifc011b5a261b7dfd13edf606c8f3cc711ca53489\nCloses-Bug: #1835161\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/dfdb5f7d5370f24d77b25d5dd81040700aa4aeb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/dfdb5f7d5370f24d77b25d5dd81040700aa4aeb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
