)]}'
{"id":"openstack%2Fpuppet-panko~674982","triplet_id":"openstack%2Fpuppet-panko~master~I97aefadedc7b8ac5e06014bd985344f2e5ec9592","project":"openstack/puppet-panko","branch":"master","topic":"fix-sphinx","hashtags":[],"change_id":"I97aefadedc7b8ac5e06014bd985344f2e5ec9592","subject":"Sync Sphinx requirement","status":"MERGED","created":"2019-08-07 01:13:42.000000000","updated":"2019-08-07 02:52:19.000000000","submitted":"2019-08-07 02:52:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"674982-1565146339291-cd85f96c","meta_rev_id":"e7140a14c8ab93e231160e4b20ab5027e9d28e30","_number":674982,"virtual_id_number":674982,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2019-08-07 02:52:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2019-08-07 02:48:12.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2019-08-07 02:48:12.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-07 02:52:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e998f22467785bb25386a62541ea5f53ceb9f27","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-07 01:13:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"124671274743afc1a24f87dcf265145b6765f0f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 01:17:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://logs.opendev.org/82/674982/1/check/build-openstack-releasenotes/1bebae4/html/ : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"e15406052d2479cdba9e39bb0b34d739e00bb8e7","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-07 02:48:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f23e5ddde6b107248d17bfe28860bd49afa59107","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 02:48:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6400b718d400d3b79e4f7052effe796f3a1c4c36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 02:52:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d7fea05bc68123f326da81ae50dba6fcb954953d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-07 02:52:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-releasenotes https://logs.opendev.org/82/674982/1/gate/build-openstack-releasenotes/57d5020/html/ : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96ad098bed3f67a2f5a4efa0e39728418bd396f2","revisions":{"96ad098bed3f67a2f5a4efa0e39728418bd396f2":{"kind":"REWORK","_number":1,"created":"2019-08-07 01:13:42.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/82/674982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/82/674982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/82/674982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/82/674982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/82/674982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/82/674982/1"}}},"commit":{"parents":[{"commit":"b89905eb63d7fbc915fc338dc02a318670b088e6","subject":"Prepare Train M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/b89905eb63d7fbc915fc338dc02a318670b088e6"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-07 01:10:09.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-08-07 01:10:09.000000000","tz":480},"subject":"Sync Sphinx requirement","message":"Sync Sphinx requirement\n\nSync sphinx dependency with global requirements. It caps python 2 since\nsphinx 2.0 no longer supports Python 2.7.\n\nChange-Id: I97aefadedc7b8ac5e06014bd985344f2e5ec9592\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/96ad098bed3f67a2f5a4efa0e39728418bd396f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/96ad098bed3f67a2f5a4efa0e39728418bd396f2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
