)]}'
{"id":"openstack%2Fpuppet-panko~676302","triplet_id":"openstack%2Fpuppet-panko~master~I1cc97eea0091812c93bdaab2be6452581b68969e","project":"openstack/puppet-panko","branch":"master","topic":"dev","hashtags":[],"change_id":"I1cc97eea0091812c93bdaab2be6452581b68969e","subject":"Add bugs url link to README","status":"ABANDONED","created":"2019-08-14 01:49:36.000000000","updated":"2019-08-15 08:28:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"612d67ea78d566bc00dc444663b13fedfcf6faff","_number":676302,"virtual_id_number":676302,"owner":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2019-08-14 01:53:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":-1,"date":"2019-08-15 08:04:56.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-14 01:53:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-08-15 08:04:56.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"37e0b28778441bd71f37ca4b20aad845d2af6d18","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-08-14 01:49:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3537b15cc3cdd0f1b2e882ebbdf8dc5ae6e077d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 01:53:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/390a8f402db44a02ae60a379522fc0e6 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"34cf4de79f7cdf6c8aeb68ce6ed4b6ce34ee02db","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-08-15 08:04:56.000000000","message":"Patch Set 1: Code-Review-1\n\nWe shouldn\u0027t do this, we are planning to move to StoryBoard which would require us to change these all over again.","accounts_in_message":[],"_revision_number":1},{"id":"855488375f94d7dff9e8a633f9762f8681e91ac1","author":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"date":"2019-08-15 08:28:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f471de0ab6f9dd771c057817f60fba6a4a0b7da","revisions":{"9f471de0ab6f9dd771c057817f60fba6a4a0b7da":{"kind":"REWORK","_number":1,"created":"2019-08-14 01:49:36.000000000","uploader":{"_account_id":23317,"name":"ShangXiao","email":"shangxiaobj@inspur.com","username":"ShangXiao"},"ref":"refs/changes/02/676302/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/02/676302/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/02/676302/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/02/676302/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/02/676302/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/02/676302/1"}}},"commit":{"parents":[{"commit":"bf360f217e6fdf35922bc78b2c9f2cd76675e9d6","subject":"Bump the openstackdocstheme extension to 1.20","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/bf360f217e6fdf35922bc78b2c9f2cd76675e9d6"}]}],"author":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2019-08-14 01:47:52.000000000","tz":-420},"committer":{"name":"ShangXiao","email":"shangxiaobj@inspur.com","date":"2019-08-14 01:49:22.000000000","tz":-420},"subject":"Add bugs url link to README","message":"Add bugs url link to README\n\nChange-Id: I1cc97eea0091812c93bdaab2be6452581b68969e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/9f471de0ab6f9dd771c057817f60fba6a4a0b7da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/9f471de0ab6f9dd771c057817f60fba6a4a0b7da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
