)]}'
{"id":"openstack%2Fpuppet-panko~684475","triplet_id":"openstack%2Fpuppet-panko~master~I0b3d7938b52f281f9a2f5dd687644e957515c449","project":"openstack/puppet-panko","branch":"master","topic":"prep-release-train-rc1","hashtags":[],"change_id":"I0b3d7938b52f281f9a2f5dd687644e957515c449","subject":"Prepare Train RC1","status":"MERGED","created":"2019-09-25 05:46:52.000000000","updated":"2019-09-27 02:00:16.000000000","submitted":"2019-09-27 02:00:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"684475-1569549616246-33733760","meta_rev_id":"46c8987eb24f2f68e399c6344da1799e4f76c8b0","_number":684475,"virtual_id_number":684475,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-09-27 02:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2019-09-27 01:49:30.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2019-09-27 01:49:30.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-27 01:49:30.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-09-27 02:00:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e8c9ce01416c2fab68f8067851aa0bfad315e22","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-09-25 05:46:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74df7e14f2de231e8a9917d1bd629285f60f623b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-25 05:55:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/aa97db72e22042c9bc5fa144c95335de : SUCCESS in 2m 12s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c323afb4de0d4bde98d4b153b8eca668 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9a3f2268151e7cf1862ad432540d3387922e401e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-09-27 01:49:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"efca35e13f9d2e762608fc3051583532d0d4e5ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 01:49:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b0578ef8e377d56d3f436f3eb3eda0217256be80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 02:00:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/68d24ea51e1f4212931dd37d913cc9ad : SUCCESS in 3m 17s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/585404fb25bb43088fef02fcfc4be690 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"0c7a80f67d21f093b265c0004bdc98310f8902be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-27 02:00:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82c46364d4c2705c1fc26dbbccfa3457def6fc32","revisions":{"82c46364d4c2705c1fc26dbbccfa3457def6fc32":{"kind":"REWORK","_number":1,"created":"2019-09-25 05:46:52.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/75/684475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/75/684475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/75/684475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/75/684475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/75/684475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/75/684475/1"}}},"commit":{"parents":[{"commit":"812fc01f2d1894a3e2811cceb803ef8032c6f12b","subject":"Merge \"Add support to configure service_token_roles in authtoken middleware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/812fc01f2d1894a3e2811cceb803ef8032c6f12b"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-25 05:45:43.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-09-25 05:45:43.000000000","tz":480},"subject":"Prepare Train RC1","message":"Prepare Train RC1\n\nUpdate the version metadata for Train milestone RC1\n\nChange-Id: I0b3d7938b52f281f9a2f5dd687644e957515c449\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/82c46364d4c2705c1fc26dbbccfa3457def6fc32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/82c46364d4c2705c1fc26dbbccfa3457def6fc32"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
