)]}'
{"id":"openstack%2Fpuppet-panko~698147","triplet_id":"openstack%2Fpuppet-panko~master~Idbe721f76c6a7b9118f305ec6d11ffb5092dd73d","project":"openstack/puppet-panko","branch":"master","topic":"prep-release-ussuri-1","hashtags":[],"change_id":"Idbe721f76c6a7b9118f305ec6d11ffb5092dd73d","subject":"Prepare Ussuri M1","status":"MERGED","created":"2019-12-10 06:25:48.000000000","updated":"2019-12-10 16:50:21.000000000","submitted":"2019-12-10 16:50:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698147-1575996621208-656d7ee0","meta_rev_id":"1cbc32e14c930f790295c6f0124cd1bc9dc4f308","_number":698147,"virtual_id_number":698147,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-10 16:50:21.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2019-12-10 16:37:02.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2019-12-10 16:37:02.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-10 16:37:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-12-10 16:50:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2c92a35488b75e006c2a402499ebed9ece5ed6d5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2019-12-10 06:25:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f35678816502beaa8ba1d75ce6a624ef187c5f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 06:35:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/5628c1e9214a4d3db9ea8d1311f67801 : SUCCESS in 2m 11s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00a3dd382fbd4625b5b3ec4d7bc1189a : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d29f635144d5256427a29ffd8f26f1c92c98f3c8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-10 16:37:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7f68b5a8c9b7b21b798efa6a372347ec9708654f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 16:37:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8496aa152975181dae0db47e9e91f3b46a9d690d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 16:50:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"34040a2b87b4dc8532f4dfcba5a14f7e88946e91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-10 16:50:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/6830233d59a6413e884f287cde75f6c5 : SUCCESS in 2m 51s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b950f8cd9e04b69b3158082a54fb72f : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6ea4f2e779608de12e2a03cc34155a4c37988ee4","revisions":{"6ea4f2e779608de12e2a03cc34155a4c37988ee4":{"kind":"REWORK","_number":1,"created":"2019-12-10 06:25:48.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/47/698147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/47/698147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/47/698147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/47/698147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/47/698147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/47/698147/1"}}},"commit":{"parents":[{"commit":"8eb1be1f1b814df8876fd622ecda66a9c9489607","subject":"Convert all class usage to relative names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/8eb1be1f1b814df8876fd622ecda66a9c9489607"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-12-10 06:24:49.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2019-12-10 06:24:49.000000000","tz":480},"subject":"Prepare Ussuri M1","message":"Prepare Ussuri M1\n\nUpdate the version metadata for Ussuri milestone M1\n\nChange-Id: Idbe721f76c6a7b9118f305ec6d11ffb5092dd73d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/6ea4f2e779608de12e2a03cc34155a4c37988ee4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/6ea4f2e779608de12e2a03cc34155a4c37988ee4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
