)]}'
{"id":"openstack%2Fpuppet-panko~707540","triplet_id":"openstack%2Fpuppet-panko~master~Iac4f37840ebdf4d717f4011138b4efa1259a6655","project":"openstack/puppet-panko","branch":"master","topic":"prep-release-ussuri-2","hashtags":[],"change_id":"Iac4f37840ebdf4d717f4011138b4efa1259a6655","subject":"Prepare Ussuri M2","status":"MERGED","created":"2020-02-13 06:25:11.000000000","updated":"2020-02-13 17:19:25.000000000","submitted":"2020-02-13 17:19:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707540-1581614365575-4edda447","meta_rev_id":"93200c3389dc9a04f000522e7f114b8a4ff5abdc","_number":707540,"virtual_id_number":707540,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 17:19:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-02-13 17:08:45.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-02-13 17:08:45.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 17:08:45.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-02-13 17:19:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fa663fa4687460524c7ee73912366550544acfe5","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-02-13 06:25:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c292ca3d3a1920ac71e151d0ffd1aa024c910e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 06:36:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/f0d529a8f4c347deb5cd13a995b79b8d : SUCCESS in 5m 30s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5daff15c6bd454895f743c7ed38add7 : SUCCESS in 7m 49s","accounts_in_message":[],"_revision_number":1},{"id":"17cc8cb5bde8ebdb826c7c120199a2aad91bd98e","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-02-13 17:08:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f7c2cfa3c9952d3e609f03366e4e9433123b62f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:08:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fef8225764c59d264b2c72dbac6e6093bd09637f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:19:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7083c414ea50d2c6943695335db347daaa8f66de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:19:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/7a79d5a938bd4c21b8971aa4c8085bc1 : SUCCESS in 3m 58s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88368f99f69e4122a4733855184b7f17 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03d57e90df32b67a03a7dd54e4e2110cce88cac9","revisions":{"03d57e90df32b67a03a7dd54e4e2110cce88cac9":{"kind":"REWORK","_number":1,"created":"2020-02-13 06:25:11.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/40/707540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/40/707540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/707540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/707540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/707540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/40/707540/1"}}},"commit":{"parents":[{"commit":"934ef674ec7a3885b5781f3a73ede419d7aa5b72","subject":"Fix stdlib version in metadata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/934ef674ec7a3885b5781f3a73ede419d7aa5b72"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-02-13 06:21:29.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-02-13 06:21:29.000000000","tz":480},"subject":"Prepare Ussuri M2","message":"Prepare Ussuri M2\n\nUpdate the version metadata for Ussuri milestone M2\n\nChange-Id: Iac4f37840ebdf4d717f4011138b4efa1259a6655\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/03d57e90df32b67a03a7dd54e4e2110cce88cac9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/03d57e90df32b67a03a7dd54e4e2110cce88cac9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
