)]}'
{"id":"openstack%2Fpuppet-panko~719503","triplet_id":"openstack%2Fpuppet-panko~master~I11665859952aa8a5d2286644fd82152392f3bc0f","project":"openstack/puppet-panko","branch":"master","topic":"bug/1872457","hashtags":[],"change_id":"I11665859952aa8a5d2286644fd82152392f3bc0f","subject":"Resolve duplicated requirement names in metadata","status":"MERGED","created":"2020-04-13 14:00:38.000000000","updated":"2020-04-13 14:18:20.000000000","submitted":"2020-04-13 14:18:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"719503-1586787500178-61dee02a","meta_rev_id":"9f99b0b9bcd0a38059652095794cff04d21d630f","_number":719503,"virtual_id_number":719503,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-13 14:18:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-13 14:10:32.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-04-13 14:10:32.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-13 14:10:32.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-13 14:18:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"35f4203d6f2955a95f8936bc1371a27619bc4dd6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-13 14:00:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2e4c632e860295df8a952c1748aa24724f57af1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 14:10:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/a24fd97bd94543ef92ac341ee1c7665d : SUCCESS in 3m 28s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/689fa41d647a4dcfb26d84ed2ba7dcf3 : SUCCESS in 6m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6553196af4c511b07387b9359ebca59d13a402c6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-13 14:10:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d6a2eea97e1ad0bdb4357b9efae6bc32443ef85f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 14:10:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6011fb2c76366daa1f5b6a94edb7a040ce9c992","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 14:18:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/a9ffc3180ad644d49ae01a90b20f3e7f : SUCCESS in 4m 18s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b0d54c67d6c4a909f1b3c5c257c4fa9 : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":1},{"id":"47c6fedbb4ed0f040be6a422a9574c0b97d0d396","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-13 14:18:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"01d68182d0f429e0d5f810f88a0cc263c69d133f","revisions":{"01d68182d0f429e0d5f810f88a0cc263c69d133f":{"kind":"REWORK","_number":1,"created":"2020-04-13 14:00:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/719503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/03/719503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/03/719503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/03/719503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/03/719503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/03/719503/1"}}},"commit":{"parents":[{"commit":"c3226aa5e874ab17cefcd9e56a38b4000b9d60e3","subject":"Prepare Ussuri M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/c3226aa5e874ab17cefcd9e56a38b4000b9d60e3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-13 14:00:28.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-13 14:00:28.000000000","tz":540},"subject":"Resolve duplicated requirement names in metadata","message":"Resolve duplicated requirement names in metadata\n\nChange-Id: I11665859952aa8a5d2286644fd82152392f3bc0f\nCloses-Bug: #1872457\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/01d68182d0f429e0d5f810f88a0cc263c69d133f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/01d68182d0f429e0d5f810f88a0cc263c69d133f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
