)]}'
{"id":"openstack%2Fpuppet-panko~722585","triplet_id":"openstack%2Fpuppet-panko~master~Idc8400ff8b344f55af356b91e7204177eac9af48","project":"openstack/puppet-panko","branch":"master","topic":"prep-release-ussuri-rc1","hashtags":[],"change_id":"Idc8400ff8b344f55af356b91e7204177eac9af48","subject":"Prepare Ussuri RC1","status":"MERGED","created":"2020-04-24 06:23:55.000000000","updated":"2020-04-25 12:54:04.000000000","submitted":"2020-04-25 12:54:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722585-1587819244668-54f1cd87","meta_rev_id":"cfe561fdcba120b2cbd6ef8c9421f6ae0b1f2f87","_number":722585,"virtual_id_number":722585,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-04-25 12:35:05.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-25 12:54:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-04-24 20:57:46.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-24 21:29:33.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-04-24 21:29:33.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-24 20:57:46.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-04-24 21:29:33.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-25 12:35:05.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-04-25 12:54:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"393676f6aacfa32314411f7bc184e55371995a19","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-24 06:23:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b29c596cb83c351fd74801d10fdb4e07960e0a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 06:39:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/a0838f0362da497fadbcb57a6a9611f9 : SUCCESS in 4m 00s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dfe544a25764c12a8d211d45840f1b6 : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e36bff1e2964405a2126b1cd9ff7537d9b3f202f","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-04-24 20:57:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"18bedca0ca64d2c03177af83abb9ac5dc78d93bb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-24 21:29:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46e81ef714e0877ae690429d14a93dcb58ae2573","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-25 12:35:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e8826de7236b79449ded73b8d68647c603fd4e1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 12:44:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/1a9861cde52847b0b8254cbf71c46df4 : SUCCESS in 2m 11s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae35e3df4ed1424b9c80c92c05292d3d : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"140cb1812e8cd1b205c0b2e9be021863ff34662b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 12:44:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10375c7dedfbc17ab8dffa7ccb3105d61e7f2a43","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 12:54:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bfe4b596145db72493e6ad5eaa207d0544a67047","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 12:54:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/11e1e2820a1940b297d9e60576b1477c : SUCCESS in 2m 09s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4d1c4b9c1054098929744a8b68276e9 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f2221ecc7842326b227af890a817c75b898f5c0","revisions":{"1f2221ecc7842326b227af890a817c75b898f5c0":{"kind":"REWORK","_number":1,"created":"2020-04-24 06:23:55.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/85/722585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/85/722585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/85/722585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/85/722585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/85/722585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/85/722585/1"}}},"commit":{"parents":[{"commit":"67fcacffb758bb1c27e8967d8cdfec41701b046c","subject":"Prepare Ussuri M3 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/67fcacffb758bb1c27e8967d8cdfec41701b046c"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-04-24 06:23:10.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-04-24 06:23:10.000000000","tz":480},"subject":"Prepare Ussuri RC1","message":"Prepare Ussuri RC1\n\nUpdate the version metadata for Ussuri milestone RC1\n\nChange-Id: Idc8400ff8b344f55af356b91e7204177eac9af48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/1f2221ecc7842326b227af890a817c75b898f5c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/1f2221ecc7842326b227af890a817c75b898f5c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
