)]}'
{"id":"openstack%2Fpuppet-panko~724014","triplet_id":"openstack%2Fpuppet-panko~stable%2Fussuri~I94e34ecb7a1c9fef5b793bfe776af3b44a5c06c8","project":"openstack/puppet-panko","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I94e34ecb7a1c9fef5b793bfe776af3b44a5c06c8","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-04-28 14:57:37.000000000","updated":"2020-04-29 16:37:32.000000000","submitted":"2020-04-29 16:34:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724014-1588178099669-bde3a662","meta_rev_id":"a7f1883d1dbf1423cad86f9f64588285f769ab77","_number":724014,"virtual_id_number":724014,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"date":"2020-04-29 14:30:40.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 16:34:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-29 01:56:00.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-28 16:47:35.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-29 01:56:00.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 16:47:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-29 01:56:00.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-04-29 14:30:40.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-04-29 16:34:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b836bba60cac75e16f7f0072d6c0659afff0171","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-28 14:57:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d3debe4e10d6df75705a70f2e50b40932e3a8fd","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-28 16:47:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1f8a76b22af044656062c89efa14697cdb2d1f96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 17:20:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f8e69b5209f4a3c9bb2f5df6889af11 : SUCCESS in 7m 02s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7c07082ee1f47a2b2eba417e7058385 : SUCCESS in 7m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7940bcedf9a4e999d26dd9c3be6a282 : SUCCESS in 6m 45s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a7657ca054584e909fe9f951121dabfc : SUCCESS in 16m 45s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/67e01abf3d1444e7b4bf701b6bf98b54 : SUCCESS in 15m 29s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/215594741ce84bc0866a77fbab7f0e66 : SUCCESS in 11m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82e70b2d242546a6a9501f2a0514a35c : SUCCESS in 13m 47s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f5bc81ea114543faad3486557728b1aa : SUCCESS in 1h 01m 07s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/84519bc0a9c54ec294882e0c49877dd1 : SUCCESS in 1h 00m 07s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/84030a360bf743fb9262e82a828dea26 : SUCCESS in 25m 02s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b2f7eae7fae747e7b5011f1f41c3a75f : FAILURE in 14m 55s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e770ae70163c4134a7d6c13f94159e4e : FAILURE in 13m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ad8b11c01bd410180becee0e2c952a0 : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":1},{"id":"fae133aa54fe65bb66c763cc1457700669ad322c","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-29 01:56:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b0e1c0bd90cb8e122b7e264df50fa5c58381ca20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 01:56:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c64dfe624597db64912cbfe240b5b1ef2c283016","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 03:01:50.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c310b349bc7f4a8c9768eb3326609aab : SUCCESS in 5m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4533bb5e8ee94be3880a243367817db2 : SUCCESS in 7m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e3b9753bf50f4fd2a2c22db7fe9f82d7 : SUCCESS in 7m 21s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a24335c0b12c429ba11f588075b7f8c5 : SUCCESS in 11m 55s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/b51212d5df0c4534a7f84b385efcf378 : SUCCESS in 14m 15s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43e60fd44570415b9d57feab102f781f : SUCCESS in 10m 49s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e5d718ad78fb4185b1d3b7ecd04f347b : SUCCESS in 45m 40s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1364bf7414544bc5890b0f3af8628713 : FAILURE in 59m 53s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bed11693ebac45bb936fb4b15842ad03 : SUCCESS in 23m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3224c3e59e344db0b6dfabc7f99c1be4 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6bf869761cf518bcc5df0283b0115ceaa9514f1f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 14:16:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"db91ad15dab613ae3ec12438d2de6de4e0a1d2ff","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 14:24:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"71c5aa6638c56db977c3b36fb1e9a8de521bdfd4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 14:30:40.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"57aab2d609cda0b41583b9b9964b80a74a5dae03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 15:33:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6f6c05477fe487b93c6f72bfbed60d1 : SUCCESS in 6m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de4393cf4448434d99389e24589a681f : SUCCESS in 5m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a960cf9af10e4f6d927db23f9eae8ef8 : SUCCESS in 6m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/fdbee6b1f6f9496bbf90804bd34b2389 : SUCCESS in 12m 35s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/84434a84a7df467f8329b666f905ec23 : SUCCESS in 12m 02s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70c98c9a5c574e5bbec6c50b54dab100 : SUCCESS in 10m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec3a244b9ebd4a768495a5c10978ab82 : SUCCESS in 12m 15s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/be74af929e97491fbab4a32173621335 : SUCCESS in 53m 07s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a36c7ee9e2934933bc6ddfe5a2bc130f : SUCCESS in 1h 00m 39s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c7737d32d7a7411892f91f2e3cb796df : SUCCESS in 18m 16s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ecc8a0b68fc34119a9d9f79e9388175c : FAILURE in 14m 27s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/083815f51aa84658ad9cd662fc26ab9c : FAILURE in 11m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d8c85cb09764ea8a04720e8e906afdd : SUCCESS in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2f0b0032449e5ec90d4608a26511c8245d399d7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 15:33:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a79d322384db6629cf4f9861846ea2fba3062099","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 16:34:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"add39e4b491f27e4d92d733fe870fd3457336860","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 16:34:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48e6f6a7bd2c4be0ac7284912ca9db7d : SUCCESS in 4m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/265d9e12ca234442a3ec2a3177a92967 : SUCCESS in 6m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/536bb932374543ecab7e91eddf88dff0 : SUCCESS in 4m 25s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b447dc2687584950ae83834be990a25d : SUCCESS in 11m 27s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/f7c78d8787f44f7c8bd83be169b09f61 : SUCCESS in 10m 27s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c38794f03932489f961c19303e694cad : SUCCESS in 7m 48s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/2a0e7a1aca5d49c0bb7c4fdb07d1813b : SUCCESS in 51m 11s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a787d16ed16943c7b43a32aa0ae53552 : SUCCESS in 55m 28s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2cefc960886647e6bd2fb52358a1ef25 : SUCCESS in 24m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ae457aa21d3b453baca92cde8a9ec821 : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":1},{"id":"94614af14887d3f55c331080262988c738e6fbd3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 16:37:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6dad44e534c341a082c8c2526affd503 : SUCCESS in 2m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90465642654994c75e625b018af62fe1e44ea1bf","revisions":{"90465642654994c75e625b018af62fe1e44ea1bf":{"kind":"REWORK","_number":1,"created":"2020-04-28 14:57:37.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/14/724014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/14/724014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/14/724014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/14/724014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/14/724014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/14/724014/1"}}},"commit":{"parents":[{"commit":"33eb9158204e3cbc23994e91d6a59e16ee25b847","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/33eb9158204e3cbc23994e91d6a59e16ee25b847"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:57:35.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:57:35.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I94e34ecb7a1c9fef5b793bfe776af3b44a5c06c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/90465642654994c75e625b018af62fe1e44ea1bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/90465642654994c75e625b018af62fe1e44ea1bf"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
