)]}'
{"id":"openstack%2Fpuppet-panko~726682","triplet_id":"openstack%2Fpuppet-panko~master~I3dad45f0ef7dfb5d8edd9fb05657f4ae59a1f40d","project":"openstack/puppet-panko","branch":"master","topic":"project-ptl-and-contrib-docs","hashtags":[],"change_id":"I3dad45f0ef7dfb5d8edd9fb05657f4ae59a1f40d","subject":"[ussuri][goal] Add CONTRIBUTING.rst","status":"MERGED","created":"2020-05-10 20:50:01.000000000","updated":"2020-05-12 12:46:04.000000000","submitted":"2020-05-12 12:46:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726682-1589287564157-e1d56cf0","meta_rev_id":"06e8b75973e3fc39c8271249366c173a5fbd0bb7","_number":726682,"virtual_id_number":726682,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-12 12:46:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-05-12 12:35:25.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-11 22:20:19.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2020-05-12 12:35:25.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 22:20:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-12 12:35:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-12 12:46:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f3a30bc684931699c2598eb98bc77fd638507aa2","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-10 20:50:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71614dc223572cb7dff284727544434e498e78c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-10 21:53:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3203930abbc14d7aa7c1f43a0a8c6f2e : SUCCESS in 4m 23s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6c0ab814eeed4e4392a2b7a6d831ab76 : SUCCESS in 45m 57s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7ae8134079dd478493208bc63fcf62fc : FAILURE in 57m 50s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/147eec0ac7c34cb7ad484c633eb70395 : SUCCESS in 17m 19s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/0edf5ae659dd4c148dd26fb29b0071ae : FAILURE in 9m 12s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4eb7f6e9eaa424785b89f1af0745ba5 : FAILURE in 6m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d524504a201e267473c8a1ec8b8742c88f4393f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-11 22:20:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"325758f513c887bf498612aa67b458945ec8fee5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-12 04:24:43.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e7fd16a0d58395dcceaa282379b8401eec2ee654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 05:17:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/635492ee34634364b5d5c940c7035358 : SUCCESS in 5m 36s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/971679c8ceea4899b8e2d9719a6a1ca3 : SUCCESS in 50m 06s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/33f9efcdc91743c2848435da2f646c55 : SUCCESS in 49m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f27cafa6c0c04cc1be30f0a7cfb711a2 : SUCCESS in 17m 15s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/1ecda20ccc0848b19d4b0a15fac602c2 : FAILURE in 8m 49s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aeb3e97e447b421cb12fdd16cab6ace2 : FAILURE in 6m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d770702f849f6413e11cc3e826b67ff0d8231819","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-12 12:35:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c5ee5c45dcc0523a737dd82eb138b1c4317bd867","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 12:35:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b6ab4abe42f6902defad2a76d6631182546a7bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 12:46:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42ca5c5942d647619177f8c941867ab9 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":1},{"id":"ae109025a2ec97e090d5b2a46886398696372316","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 12:46:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"572d50f413b22aa8eab9259c3f4ab700e3299f12","revisions":{"572d50f413b22aa8eab9259c3f4ab700e3299f12":{"kind":"REWORK","_number":1,"created":"2020-05-10 20:50:01.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/82/726682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/82/726682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/82/726682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/82/726682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/82/726682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/82/726682/1"}}},"commit":{"parents":[{"commit":"36b1e4416bd9c620a4fbf77ced4a9029e6d7e0a9","subject":"Merge \"Use anchor to require necessary packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/36b1e4416bd9c620a4fbf77ced4a9029e6d7e0a9"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-10 20:49:53.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-10 20:49:53.000000000","tz":120},"subject":"[ussuri][goal] Add CONTRIBUTING.rst","message":"[ussuri][goal] Add CONTRIBUTING.rst\n\nStory: 2007236\nTask: 38552\n\nDepends-On: https://review.opendev.org/#/c/726509/\nChange-Id: I3dad45f0ef7dfb5d8edd9fb05657f4ae59a1f40d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/572d50f413b22aa8eab9259c3f4ab700e3299f12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/572d50f413b22aa8eab9259c3f4ab700e3299f12"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
