)]}'
{"id":"openstack%2Fpuppet-panko~728199","triplet_id":"openstack%2Fpuppet-panko~stable%2Fussuri~Id6796d02a482f0ca3bb2fb85761208199e10dc8d","project":"openstack/puppet-panko","branch":"stable/ussuri","hashtags":[],"change_id":"Id6796d02a482f0ca3bb2fb85761208199e10dc8d","subject":"Switch to Ussuri","status":"MERGED","created":"2020-05-14 18:07:11.000000000","updated":"2020-05-18 06:53:51.000000000","submitted":"2020-05-18 06:53:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728199-1589784831539-bb5f142c","meta_rev_id":"b8b8fc37564d20770f9f68a14868f3fa4f85d887","_number":728199,"virtual_id_number":728199,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 06:53:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-05-18 05:58:22.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-05-15 22:46:06.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-05-18 05:58:22.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 22:46:06.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-18 05:58:22.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-05-18 06:53:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e92e2eaeb61a44a5e42122e7d24f84d860be0df","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-14 18:07:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"310d27a5199b9f8dda7c49263fa5c259141a1fe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 19:23:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04a9133633264650836a8d1698478319 : SUCCESS in 8m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b552e532438948538cadc5c92e2556ce : SUCCESS in 5m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e601246200414de1863fe5f7f06616bf : SUCCESS in 4m 58s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b270462379564feb9630d9f5aaa29229 : SUCCESS in 15m 35s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/5794f64a06a0467fa0dc8a23cbe0b0e0 : SUCCESS in 14m 31s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/313ed29668cf45a084ff4f29b8e9fee1 : SUCCESS in 11m 02s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ebf730ce8ef4a00a6d7e79da2ead37e : SUCCESS in 12m 11s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/daba87b1bc6e45dfb1e1e6d6a717d9d0 : SUCCESS in 45m 28s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6e7fccdf751f4b2d821c8f10931812b4 : SUCCESS in 1h 01m 34s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/aff7895bdb9a4abb9f27e16c6954583d : SUCCESS in 19m 01s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/4f61131240044c759f0e4e13e2ca7d04 : FAILURE in 13m 46s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a01312159534413ab75305e27592b5f : FAILURE in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ef499c1674f0c0438148f09916fb93077e80990","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-15 22:46:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f282077e40e575d76533aec47faf73547f501680","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-05-18 05:58:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b1652d6489615620ee5508e995f443094a4b72a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 05:58:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f0017e49ee475fffe3c6a20e59f0ca4e6d01bcd3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 06:53:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b8b8fc37564d20770f9f68a14868f3fa4f85d887","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 06:53:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e32e2d4810be4b908338ca85b482c3ea : SUCCESS in 5m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afa29fac840249c088a4ee7898c93057 : SUCCESS in 4m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfb053a0cd9d4b58a9caccae53d70511 : SUCCESS in 5m 56s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/17b1255f41f84cc1871fd811ac9ecb4a : SUCCESS in 15m 15s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/218a1bdbc5504dffb9796e6aa4bc4ff9 : SUCCESS in 11m 28s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0905eb105a8343a89e893c0aad9d9569 : SUCCESS in 7m 57s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0e00c5d906174005a8400300e93562c5 : FAILURE in 39m 23s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/42b869c909894508b3a1ebe4f1d53adc : SUCCESS in 54m 23s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/30f9f7d361d242fa9bcff0622960b623 : SUCCESS in 20m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79b30f16540bf3e69969b6562d7187bb261c12d2","revisions":{"79b30f16540bf3e69969b6562d7187bb261c12d2":{"kind":"REWORK","_number":1,"created":"2020-05-14 18:07:11.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/99/728199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/99/728199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/99/728199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/99/728199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/99/728199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/99/728199/1"}}},"commit":{"parents":[{"commit":"90465642654994c75e625b018af62fe1e44ea1bf","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/90465642654994c75e625b018af62fe1e44ea1bf"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-14 18:07:02.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-05-14 18:07:02.000000000","tz":120},"subject":"Switch to Ussuri","message":"Switch to Ussuri\n\nChange-Id: Id6796d02a482f0ca3bb2fb85761208199e10dc8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/79b30f16540bf3e69969b6562d7187bb261c12d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/79b30f16540bf3e69969b6562d7187bb261c12d2"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
