)]}'
{"id":"openstack%2Fpuppet-panko~740130","triplet_id":"openstack%2Fpuppet-panko~master~I8ae2e126048c130f7c498d5ec0e790bc0dd42c1e","project":"openstack/puppet-panko","branch":"master","topic":"at-interface","hashtags":[],"change_id":"I8ae2e126048c130f7c498d5ec0e790bc0dd42c1e","subject":"Add support for the interface parameter in authtoken middleware","status":"MERGED","created":"2020-07-08 23:37:39.000000000","updated":"2020-07-09 09:00:00.000000000","submitted":"2020-07-09 08:57:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"740130-1594285046567-72fcc681","meta_rev_id":"5edbd684d51b69e94940d8a2dcd37a76c6bef8f8","_number":740130,"virtual_id_number":740130,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-09 08:57:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-07-09 02:02:01.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-07-09 07:46:31.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2020-07-09 07:46:31.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-09 02:02:01.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-07-09 07:46:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-07-09 08:57:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e7e6c70cdf9d8a732d09497b1c07e3bc675b513","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-08 23:37:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0462b4f65f255c4732b0d2cfe92fe919ead557b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 00:40:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e78f1fa65e54728bc8b4d56830f4cf1 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6fb431ea12b4350aa6a97cfa106b18e : SUCCESS in 5m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4176a9e6aef74270941f843e0eb2fda2 : SUCCESS in 5m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f71ab38b0a514683884e0bba89ed4341 : SUCCESS in 12m 54s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/66fa8e9293304cc2b31dee0d529e34a3 : SUCCESS in 15m 39s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8adc64e8455a4357ab8d5e1bc45de270 : SUCCESS in 13m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/480dfdf1946146cba0b5eb9ba6021b16 : SUCCESS in 12m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/ccdbc174a93a4eb1a5298384b4a7e3dc : SUCCESS in 1h 00m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fae3abff3a9f48eeb93708b1be5bbed7 : SUCCESS in 58m 00s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c6cf64fc9e954edda8d21eb1dca3e0b9 : SUCCESS in 17m 08s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/75da3dae448a407a9845affb68387bab : FAILURE in 10m 04s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7a897ca2b35498ca310983b7897afdc : FAILURE in 11m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c375a782c6546f1a20125ddbfc7efa0 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0a835d5c6ed5d00dece9cec6a8c9e5c5d4326430","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-07-09 02:02:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a8f8c8eac08a23c00d12436b76c4347dd6eb864","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-09 07:46:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bf99fd97435c03e2bb3c4c947a964e3eb35bc99f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 07:46:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e072be9117087f5dca61fcfa23dec5c6a326e3c4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 08:57:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1a8b405745b986db6693f7808afe226d1cf4f581","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 08:57:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afe646c4efaf438aa542947a59aa8a33 : SUCCESS in 5m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d462cb4330d431086b0172d9a739f53 : SUCCESS in 7m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3115270ef014fe09de53bb66133fd9a : SUCCESS in 5m 26s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/f44517c53e10450b9b6034c2ba9d3442 : SUCCESS in 11m 58s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/568b7d758d694185baa5642d0cdd0559 : SUCCESS in 14m 22s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fbe0031377443b086eec6ac67a5e358 : SUCCESS in 13m 30s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/359e2c4b805340fa8adcc68b19b56a64 : SUCCESS in 1h 00m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a27c28fb06b34782b6ac701b82c2bf20 : SUCCESS in 1h 05m 47s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/73ce1a8b2ffd44a79117f6cbaf171b77 : SUCCESS in 21m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7defe08997b4048824296398d06633b : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"783352be8b2da011d8c23f1e7c21b8a61043bc89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-09 09:00:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29144813700947758490ef7bb18110f1 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba317b7ef3c6212cc60473457dae94ca828959bf","revisions":{"ba317b7ef3c6212cc60473457dae94ca828959bf":{"kind":"REWORK","_number":1,"created":"2020-07-08 23:37:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/740130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/30/740130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/30/740130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/30/740130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/30/740130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/30/740130/1"}}},"commit":{"parents":[{"commit":"a406170d803da0fe4dbb9913f92322e7b951eba3","subject":"Prepare Victoria M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/a406170d803da0fe4dbb9913f92322e7b951eba3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-08 23:37:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-08 23:37:24.000000000","tz":540},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: I8ae2e126048c130f7c498d5ec0e790bc0dd42c1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ba317b7ef3c6212cc60473457dae94ca828959bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ba317b7ef3c6212cc60473457dae94ca828959bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
