)]}'
{"id":"openstack%2Fpuppet-panko~746240","triplet_id":"openstack%2Fpuppet-panko~master~Icb3e02dc92e430747e276e935440044a9cc53c95","project":"openstack/puppet-panko","branch":"master","topic":"expirer-maxdelay","hashtags":[],"change_id":"Icb3e02dc92e430747e276e935440044a9cc53c95","subject":"Add maxdelay parameter for panko-expirer cron job","status":"MERGED","created":"2020-08-13 23:34:40.000000000","updated":"2020-08-17 16:52:53.000000000","submitted":"2020-08-17 16:50:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746240-1597683015142-7da3442f","meta_rev_id":"301364cc808aedd4c9a64f532fcedd7504550b3a","_number":746240,"virtual_id_number":746240,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-17 16:50:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-17 15:22:30.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-08-14 09:03:17.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-17 15:22:30.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 09:03:17.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-17 15:22:30.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-17 16:50:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2446c9a37ddcd0ac50689da5838fce7ffd94ddc6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-13 23:34:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96d4f206aa1a0d36378593d34a789cd72e5ada5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 00:40:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa4df538061745a7a9fe43397976e4ff : SUCCESS in 4m 38s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d032ce0a76da4c6099a69dffa230736f : FAILURE in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b7884c7a75c34292ad9a58d7f038f9f4 : FAILURE in 4m 54s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bcbcc89bfea74ff9a55dbb7ec949395c : FAILURE in 12m 26s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e82b307221e545b8b5a99da73f139b0b : FAILURE in 12m 30s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/933391e235dd49cfbaf2119fc4592997 : FAILURE in 10m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82611f25514f427fa495e603375c4c6a : FAILURE in 9m 45s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c8ce05dd30b2456c8e9cb8f688548c3c : SUCCESS in 48m 03s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/66b6a433cece447eabae33016fefb9d4 : SUCCESS in 1h 03m 03s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2fac5b3ceeb04d1fb42ae24775aaff56 : SUCCESS in 20m 32s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9bf168158b54435fb72767247d51d517 : FAILURE in 10m 48s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ead64fab224b4909bfa7dbdea3ef4ab3 : FAILURE in 8m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf73b3f87ab14b928ea5118cad9cf788 : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7babb274a20223edd96ab21b04902011e7eaf9b3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-14 01:57:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f867c9368656fbd4496336dac26156bf6b3d76d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 03:21:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/700968d57ab146a8a83631c61600c8f3 : SUCCESS in 6m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9a0d40302154d339b975fdebb851c6d : SUCCESS in 5m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2609ff8d84a5471aadc414b81f8b4df8 : SUCCESS in 6m 13s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/a51cecb442934b068d1ed092e9d6c038 : FAILURE in 12m 19s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e7d769c4ead24ab2a6cbbc774ceddf1d : FAILURE in 11m 02s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/613fdae842e34bbc97996b515580c019 : FAILURE in 7m 48s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9305dcf2dae44321b5954b325388c0e6 : FAILURE in 6m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/a89268e0acc14accb6cbfc02177d7f77 : SUCCESS in 48m 54s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fd8c8cb3862945108f38375bbe5b0d60 : SUCCESS in 1h 17m 41s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0e9aac46828348ff85843fe1cee01513 : SUCCESS in 17m 37s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ad30f04cf9b5400a86e58b21acb4e0d7 : FAILURE in 9m 11s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb2d8957021146a49e121afca5df2d76 : FAILURE in 7m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/308fac9fee594d84a84cce436d9931ff : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"6a7e0c813b10fdab98bd2f6e7058d7fcb80fd7c9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-14 04:30:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"57f2c3bddce5390e8ead13654a0ae94a9d22360f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 05:53:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15191589131e45d8aeb48b69bfdd1f91 : SUCCESS in 4m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e28c1b961fb44ab4a71d344f9aa1e5c8 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fe8d245accfc4d928e3e44234f06e491 : SUCCESS in 4m 25s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8da9c1cc55cb463ba39c067090b79bc7 : FAILURE in 8m 51s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f20fa43434284b888380f4b6eded295c : FAILURE in 15m 26s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b3ee036fa52455f8b0947cf71b112f4 : FAILURE in 8m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/382db5157b43474cadc4c78f1a81ca7e : FAILURE in 6m 17s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/488d395c196c446b890679601fe7a568 : SUCCESS in 58m 15s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a9dc422b16d14e3d88b8ea8682dca2bd : SUCCESS in 1h 17m 29s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b9c9a102e4ba496f93a49a7f00a3e8ef : SUCCESS in 18m 29s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/ce811c9c69cd46099169a9fc389680b5 : FAILURE in 7m 26s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/511a26d6588642fdadaa9daa287b9b86 : FAILURE in 5m 04s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/281e39642e634194b8993911c4bee197 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":3},{"id":"b61b9b9808204d0620a7b5d10ed5b08ccc80932a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-14 06:50:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ba0bffe87800234e4445ef91e87c8732bd655b5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 07:43:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7a318fb28e594bc5b2f51956ce5bfa17 : SUCCESS in 4m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4dc8a44b299f483ca16cd9112387b3ad : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/558250c535764e4b82396c43b8b5b7c3 : SUCCESS in 5m 45s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6374d63becc6446a977affdb1339e20f : SUCCESS in 14m 21s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3c8d2cc1c7ee440fb827671c8836ce22 : SUCCESS in 12m 18s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5347fb9233cc4977bb7afb48437fdd03 : SUCCESS in 9m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/931c04968e7c4c77bc27418078337b20 : FAILURE in 11m 16s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5e203b4bf9494ba286c5c52a56531374 : SUCCESS in 37m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2d36416376eb4879920af5bf4cf63905 : SUCCESS in 48m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7b34a0a9c30742e79be003d0ac7aeeba : SUCCESS in 14m 13s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/70d1b28812784926916fd0ba31aa56fe : FAILURE in 8m 50s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b28736eca3c488dadf5bb12c852714c : FAILURE in 9m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fd6b9f8b3d2493688c80df1d6526ce3 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":4},{"id":"66097a6fabb2cb442c75a066bb58d14e3e5405c7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-14 09:03:17.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e708bc9b903419508982a30078f57190513e251b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-17 15:22:30.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"61dc8516e324f4216d9c2b8d70e600aa3043ab52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 15:23:23.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6357ae9ef9180b60ee8aa862e291ea71b0625568","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 16:50:14.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c2481d69e0d4760a429a5838c7eb3d4 : SUCCESS in 5m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/01d24d0b1f7e4a9c8c49890bc163d020 : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d53774d6558f46c7829d25c0fd76f8c5 : SUCCESS in 5m 19s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/85ee77ad3e054df38a0d9bb64b7499a3 : SUCCESS in 15m 08s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/347d0f2e4c924a0f94d9f3a5baeb4174 : SUCCESS in 12m 42s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7904a8461eed49428bb3bffaf6c9b831 : SUCCESS in 10m 12s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/14b702dfe9314d5d81661684e084c98a : SUCCESS in 50m 10s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/57d5d87d3f0a4995aece396a028c878e : SUCCESS in 1h 17m 14s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d8f499dcf034473183642251136553cf : SUCCESS in 19m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf2bb0fd26ac4943a145e05e10b4ddde : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":4},{"id":"6403a71369e44256997eca815872bb500e73be2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 16:50:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"964801296be68a5ad1b20810e9dede178bd19bc6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 16:52:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bfd27e8e59d4b3d89bf72e55b97ee30 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"310d39a615c049bdb04a11da260e99ca33ab15ca","revisions":{"366a3093f8301a3919628fb4790c0f7f26745143":{"kind":"REWORK","_number":1,"created":"2020-08-13 23:34:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/746240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/40/746240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/1"}}},"commit":{"parents":[{"commit":"9f18873519fe1afe35afe244d295aa8429b999dc","subject":"Use common class to manage apache service in beaker job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/9f18873519fe1afe35afe244d295aa8429b999dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-13 23:31:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-13 23:34:32.000000000","tz":540},"subject":"Add maxdelay parameter for panko-expirer cron job","message":"Add maxdelay parameter for panko-expirer cron job\n\nThis patch introduced a new parameter, panko::expirer::max_delay, which\nconfigures randomized delay in panko-expirer execution, so that we can\navoid collisional execution of the command from multiple nodes.\n\nChange-Id: Icb3e02dc92e430747e276e935440044a9cc53c95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/366a3093f8301a3919628fb4790c0f7f26745143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/366a3093f8301a3919628fb4790c0f7f26745143"}]},"branch":"refs/heads/master"},"00c639e1070a5f6be822ebaa1f21347e3fc71722":{"kind":"REWORK","_number":2,"created":"2020-08-14 01:57:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/746240/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/40/746240/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/2"}}},"commit":{"parents":[{"commit":"9f18873519fe1afe35afe244d295aa8429b999dc","subject":"Use common class to manage apache service in beaker job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/9f18873519fe1afe35afe244d295aa8429b999dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-13 23:31:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-14 01:57:46.000000000","tz":540},"subject":"Add maxdelay parameter for panko-expirer cron job","message":"Add maxdelay parameter for panko-expirer cron job\n\nThis patch introduced a new parameter, panko::expirer::max_delay, which\nconfigures randomized delay in panko-expirer execution, so that we can\navoid collisional execution of the command from multiple nodes.\n\nChange-Id: Icb3e02dc92e430747e276e935440044a9cc53c95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/00c639e1070a5f6be822ebaa1f21347e3fc71722"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/00c639e1070a5f6be822ebaa1f21347e3fc71722"}]},"branch":"refs/heads/master"},"bdbe602232f176fd06a37c4e015e982f6d25e0bc":{"kind":"REWORK","_number":3,"created":"2020-08-14 04:30:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/746240/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/40/746240/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/3"}}},"commit":{"parents":[{"commit":"9f18873519fe1afe35afe244d295aa8429b999dc","subject":"Use common class to manage apache service in beaker job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/9f18873519fe1afe35afe244d295aa8429b999dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-13 23:31:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-14 04:30:31.000000000","tz":540},"subject":"Add maxdelay parameter for panko-expirer cron job","message":"Add maxdelay parameter for panko-expirer cron job\n\nThis patch introduced a new parameter, panko::expirer::max_delay, which\nconfigures randomized delay in panko-expirer execution, so that we can\navoid collisional execution of the command from multiple nodes.\n\nChange-Id: Icb3e02dc92e430747e276e935440044a9cc53c95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/bdbe602232f176fd06a37c4e015e982f6d25e0bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/bdbe602232f176fd06a37c4e015e982f6d25e0bc"}]},"branch":"refs/heads/master"},"310d39a615c049bdb04a11da260e99ca33ab15ca":{"kind":"REWORK","_number":4,"created":"2020-08-14 06:50:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/746240/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/40/746240/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/40/746240/4"}}},"commit":{"parents":[{"commit":"9f18873519fe1afe35afe244d295aa8429b999dc","subject":"Use common class to manage apache service in beaker job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/9f18873519fe1afe35afe244d295aa8429b999dc"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-13 23:31:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-14 06:50:09.000000000","tz":540},"subject":"Add maxdelay parameter for panko-expirer cron job","message":"Add maxdelay parameter for panko-expirer cron job\n\nThis patch introduced a new parameter, panko::expirer::max_delay, which\nconfigures randomized delay in panko-expirer execution, so that we can\navoid collisional execution of the command from multiple nodes.\n\nChange-Id: Icb3e02dc92e430747e276e935440044a9cc53c95\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/310d39a615c049bdb04a11da260e99ca33ab15ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/310d39a615c049bdb04a11da260e99ca33ab15ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
