)]}'
{"id":"openstack%2Fpuppet-panko~746314","triplet_id":"openstack%2Fpuppet-panko~master~I00518ee991df9ab8fbfcb3309413939c500849f7","project":"openstack/puppet-panko","branch":"master","topic":"expirer-ensure","hashtags":[],"change_id":"I00518ee991df9ab8fbfcb3309413939c500849f7","subject":"Rename the parameter to define the state of panko-expirer cron job","status":"MERGED","created":"2020-08-14 10:17:12.000000000","updated":"2020-08-21 05:07:01.000000000","submitted":"2020-08-21 05:05:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"746314-1597986348507-14551777","meta_rev_id":"249493f2c4f1cfe5cd21c39662bbfe7bfbfd252e","_number":746314,"virtual_id_number":746314,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-08-21 03:01:46.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 05:05:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-17 15:22:35.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-08-20 14:55:46.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-08-20 14:55:46.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-14 20:59:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-17 15:22:35.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-20 14:55:46.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-21 05:05:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a480c5dc5c8aed3c34176ad0441da98d8187f6d0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-14 10:17:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3360c4043f5d9e975907003b2e54b67e59bb0b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 11:50:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4087691351fa4908857ee6102bea0ad4 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7ceb62e67c544aebe95cd6442fd1f7d : SUCCESS in 4m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f165444a6ea48c9abfdd7094652cba2 : SUCCESS in 4m 05s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3a83974e2abb47cf89343251aaa40491 : SUCCESS in 13m 06s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d068ecc601a3406ea6887327d7b855c4 : SUCCESS in 11m 20s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98c4a08043464363922356dffeca03e7 : SUCCESS in 9m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dcc6cdab624349caadcde2fc9d6e0d64 : FAILURE in 7m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/51644a61957249dc90afde84365a2148 : SUCCESS in 54m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/41be72c4fc9a4be6ba0b2fff2f444ebb : SUCCESS in 1h 25m 00s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b3d64912d7e44c2f9e0969ed66eda16d : SUCCESS in 17m 31s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/3518cc668f7545d581c846b72ee3dce3 : FAILURE in 11m 19s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4b8de4cf92347c688f0ce03a9dc33d2 : FAILURE in 6m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2d520738e444c3186dc03d51191f50a : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":1},{"id":"63b6898d3b7d0e3b8b15375030ec6bbce6aad44a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-14 12:22:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7751bf24d5bcc0c0f3cae51e5b7fd1854aa50bfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 13:26:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2a8dc1eb96744dea18aefac7f94cec5 : SUCCESS in 4m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1482b77c43154a27b262ba52d02c31d5 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6797b494166b418382629ebf22c76dcc : SUCCESS in 4m 24s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/53dedf520ea24685817dede7340e0aed : SUCCESS in 11m 33s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d3000314092b4588825f0e43a9845fa6 : SUCCESS in 11m 37s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b88237b57e14b2eb6b8dd537805555e : SUCCESS in 9m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba5e612587e14993842fe706b5c58294 : FAILURE in 6m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/4ba9cee87d4f4af892240da23dda4de1 : SUCCESS in 48m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/911be0583f5c4aad94180062d587a94d : SUCCESS in 1h 02m 02s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9afcfe3be69047e09a3001d9f0a39fc7 : SUCCESS in 17m 14s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/6ff12ad6dbe0420cb5a730937ac1f451 : FAILURE in 7m 17s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8839b13d3aa14ec8a54f4e7f8129fb6e : FAILURE in 5m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e2c82687ed1497aba7920bd0db63a5c : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"bfb319fa9a4dd6addaf4c0d149ea6f19be1847a4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-14 20:59:58.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ea0a946ce894ec56b5860e1fb98f8b71408dbddf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-16 23:00:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"00642d317304d9f4a7bf46ab855486fadf484b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-17 00:43:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a3917718ff8460d92853fd558f00e40 : SUCCESS in 6m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b33f91265bb743a7ace01c7990496de8 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a69d9904b972405887bc1291f91e7ef7 : SUCCESS in 5m 39s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2cb68838dd8443d18c34becdc48d1b6d : SUCCESS in 9m 32s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c057714c67c84260a650f0e9e43f654b : SUCCESS in 11m 13s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9233f46935554835bd42c675ef58984d : SUCCESS in 7m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/00a8693c824b4836a4a98f4a63e4d3bd : FAILURE in 5m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/02fa0035f081416dbab0ace05a543d81 : SUCCESS in 50m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a2e80f308d0e4a5eb85d4ab7f67d2743 : SUCCESS in 1h 41m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/082b63e2fdb749e2aebba89bb9d0a970 : SUCCESS in 15m 50s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/497429d766704b82acb7b68dc4bd1be9 : FAILURE in 9m 06s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c69e8484842d47ce92f876e52abf320f : FAILURE in 6m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cfd6ce214494f1ba1fbb64c40d322f2 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":3},{"id":"7f0fba34d690bb161dca124492f9f5b8e2df2008","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-17 08:25:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e449dd36817b2004326d720cb5b6daf973e0f1a6","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-17 15:22:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4bfca9545f3936c38e1701d7ee8edc63c0492089","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-20 14:55:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c176a0d0c04031832a69b8c2cd9050d6eeeae10e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 14:55:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a2fbcd9eefd965f774537d440aaf2fa5f931aafc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-20 16:08:13.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d706b96b1ab4e69a39eefd60ebce319 : RETRY_LIMIT in 1m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c11154244f934c4481e7b1b6c6bfab4c : RETRY_LIMIT in 1m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47d1dbb02b394c3f893ec8fa5a906393 : RETRY_LIMIT in 1m 36s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4c5bcfacfd43434dbfaf91d1a8595d96 : SUCCESS in 12m 34s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/37a0057dfe094deda3fab1e5970f0b9d : SUCCESS in 13m 00s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74c57fcf782347af80552a3e2cdd992c : RETRY_LIMIT in 1m 59s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5528df1978a64782893e807a53b2797d : RETRY_LIMIT in 1m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/89cc6a3067354fc9911ec63fe6e41cef : SUCCESS in 1h 04m 15s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/3a3ef8d3f6f44731acca99f3077ca2c8 : SUCCESS in 18m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/280e326368584f179eadeca07938c140 : RETRY_LIMIT in 1m 44s","accounts_in_message":[],"_revision_number":3},{"id":"b6e23712fb33a2be70d8796234da8a2a13644f55","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-21 03:01:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cf467478dc8d6b07df30e191a0d3bbbce07182eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 03:49:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e86d9a1336f4c74b95257012d8a7221 : SUCCESS in 4m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7abfab02f0f14949997733747523b487 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3dfb27f77a074022b2ecc630792d045b : SUCCESS in 4m 39s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4c52871f25584855b54511dbe9d23222 : SUCCESS in 9m 32s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/360810edae324db0bb37a8e6068fee46 : SUCCESS in 10m 01s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87f1517eb70f49319582cdc26a4d27c4 : SUCCESS in 7m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5eef76efc469417996546e295c7cd6e3 : FAILURE in 6m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5982e9ff4a5e482cb051f3928a94bdc5 : SUCCESS in 38m 46s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dea09feec8694febb4eaa45b42bab3f7 : SUCCESS in 46m 00s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/545a1343a4a1466f99d3b74887d57eaa : SUCCESS in 13m 15s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/22c6591a851d4b0d8e8d087fd85d48a8 : FAILURE in 8m 01s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/732c8a41c8f6451586403992cfc46bf8 : FAILURE in 5m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84379d7d905149b582717e06871138c7 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":3},{"id":"788619ff6e0e4b3b22be87bb4d71aa7adb0dfdff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 03:50:05.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"141cb4958a5731e9e98917cdca4ce294c083cfda","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 05:05:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9623612785e27408c0efd9ae65d08f7f6baa000a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 05:05:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5adcde7db5d148dea58685fc8d442c33 : SUCCESS in 3m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d0743d9b6a44316b79fb5f0946263c1 : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59853e600a9f403d80dce2d13e572176 : SUCCESS in 4m 38s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/707376cb8ad84b4cb7be5724b5df397e : SUCCESS in 11m 27s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3e92b70d275b488eb73fad2c8f4f8aec : SUCCESS in 10m 22s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1d6b0fd377f4a01bb4a843cdab301d0 : SUCCESS in 9m 39s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/fcd955718afd4621b6e4eb82726082d8 : SUCCESS in 46m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aaede0227b104a7e8939fb9801785ba9 : SUCCESS in 1h 13m 24s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4567582490eb4a859d47a1e6586f91b4 : SUCCESS in 24m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0d213184f1141459cbd19509eb02b7f : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":3},{"id":"64aa3214eb8c8ccb1b4e317b51e87fc005abdd4f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 05:07:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b32cab2e79f4529a996b7cfb9d91db7 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4184e32676123fc00fd22e1334de2d8cb499c29c","revisions":{"dab3cbf90ff9797d060ded248a27ede38f6d4f7e":{"kind":"REWORK","_number":1,"created":"2020-08-14 10:17:12.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/746314/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/14/746314/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/14/746314/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/14/746314/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/14/746314/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/14/746314/1"}}},"commit":{"parents":[{"commit":"310d39a615c049bdb04a11da260e99ca33ab15ca","subject":"Add maxdelay parameter for panko-expirer cron job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/310d39a615c049bdb04a11da260e99ca33ab15ca"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-14 10:15:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-14 10:15:49.000000000","tz":540},"subject":"Rename the parameter to define the state of panko-expirer cron job","message":"Rename the parameter to define the state of panko-expirer cron job\n\nThis patch replaces the panko::expirer::enable_cron parameter by\npanko::expirer::ensure parameter, so that the parameter name shows\nhow the parameter actually behaves more clearly.\n\nChange-Id: I00518ee991df9ab8fbfcb3309413939c500849f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/dab3cbf90ff9797d060ded248a27ede38f6d4f7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/dab3cbf90ff9797d060ded248a27ede38f6d4f7e"}]},"branch":"refs/heads/master"},"181dc74b2dfe2d9693b197261b4dbbdc898beaeb":{"kind":"REWORK","_number":2,"created":"2020-08-14 12:22:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/746314/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/14/746314/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/14/746314/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/14/746314/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/14/746314/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/14/746314/2"}}},"commit":{"parents":[{"commit":"310d39a615c049bdb04a11da260e99ca33ab15ca","subject":"Add maxdelay parameter for panko-expirer cron job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/310d39a615c049bdb04a11da260e99ca33ab15ca"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-14 10:15:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-14 12:21:52.000000000","tz":540},"subject":"Rename the parameter to define the state of panko-expirer cron job","message":"Rename the parameter to define the state of panko-expirer cron job\n\nThis patch replaces the panko::expirer::enable_cron parameter by\npanko::expirer::ensure parameter, so that the parameter name shows\nhow the parameter actually behaves more clearly.\n\nChange-Id: I00518ee991df9ab8fbfcb3309413939c500849f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/181dc74b2dfe2d9693b197261b4dbbdc898beaeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/181dc74b2dfe2d9693b197261b4dbbdc898beaeb"}]},"branch":"refs/heads/master"},"4184e32676123fc00fd22e1334de2d8cb499c29c":{"kind":"REWORK","_number":3,"created":"2020-08-16 23:00:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/746314/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/14/746314/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/14/746314/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/14/746314/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/14/746314/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/14/746314/3"}}},"commit":{"parents":[{"commit":"310d39a615c049bdb04a11da260e99ca33ab15ca","subject":"Add maxdelay parameter for panko-expirer cron job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/310d39a615c049bdb04a11da260e99ca33ab15ca"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-14 10:15:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-16 22:59:53.000000000","tz":540},"subject":"Rename the parameter to define the state of panko-expirer cron job","message":"Rename the parameter to define the state of panko-expirer cron job\n\nThis patch replaces the panko::expirer::enable_cron parameter by\npanko::expirer::ensure parameter, so that the parameter name shows\nhow the parameter actually behaves more clearly.\n\nChange-Id: I00518ee991df9ab8fbfcb3309413939c500849f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/4184e32676123fc00fd22e1334de2d8cb499c29c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/4184e32676123fc00fd22e1334de2d8cb499c29c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
