)]}'
{"id":"openstack%2Fpuppet-panko~752155","triplet_id":"openstack%2Fpuppet-panko~master~Ide863b35399645bb4951a767e14219fa3219290b","project":"openstack/puppet-panko","branch":"master","topic":"add_openstackdocs_auto_name","hashtags":[],"change_id":"Ide863b35399645bb4951a767e14219fa3219290b","subject":"Set openstackdocs_auto_name to use \u0027project\u0027 as name","status":"MERGED","created":"2020-09-16 03:08:14.000000000","updated":"2020-09-16 03:52:07.000000000","submitted":"2020-09-16 03:51:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752155-1600228263820-8dffc5fa","meta_rev_id":"b5f7a50f6c74703d2a3b14bd80c325ef0b040a02","_number":752155,"virtual_id_number":752155,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-16 03:51:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-09-16 03:46:40.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-09-16 03:46:40.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-16 03:51:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"065aa38803d338d36417e670f9f1eb977a2c3485","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-09-16 03:08:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01aa59afbe4f1998846ff2eaf6f4ac153fac95b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 03:12:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3f84618bd6d4a36bd118f1ab6c6cb6d : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c930320049202d3cfb0a0925747781a70e22fa04","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-09-16 03:46:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f0b32ff4331d6bdefb9ba485a86c1e1b5c51bb90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 03:46:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"64e65017457b7d986e4deb08d214c808f5805bc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 03:51:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64ad0b9c35b4431b98e3fbe6d0783668 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0e46512bc8ca86e0723b39f52c864ddd9f61b757","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 03:51:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"60f3ee8883ad9ac730ab743773283458dfb99652","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 03:52:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8e3116fddec42a9a847107fab90cb6b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c2d356aeac2cbeaa516a429b15b9d58d3464fe2","revisions":{"0c2d356aeac2cbeaa516a429b15b9d58d3464fe2":{"kind":"REWORK","_number":1,"created":"2020-09-16 03:08:14.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/55/752155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/55/752155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/55/752155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/55/752155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/55/752155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/55/752155/1"}}},"commit":{"parents":[{"commit":"0d54ee4287f4505f66fcd1c8c38003b934431a75","subject":"Merge \"Add Puppet Litmus\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/0d54ee4287f4505f66fcd1c8c38003b934431a75"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-09-16 03:05:05.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-09-16 03:05:05.000000000","tz":480},"subject":"Set openstackdocs_auto_name to use \u0027project\u0027 as name","message":"Set openstackdocs_auto_name to use \u0027project\u0027 as name\n\nChange-Id: Ide863b35399645bb4951a767e14219fa3219290b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/0c2d356aeac2cbeaa516a429b15b9d58d3464fe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/0c2d356aeac2cbeaa516a429b15b9d58d3464fe2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
