)]}'
{"id":"openstack%2Fpuppet-panko~753550","triplet_id":"openstack%2Fpuppet-panko~master~I7ca0ab9e42352645e463c49f8ce58793115543c0","project":"openstack/puppet-panko","branch":"master","hashtags":[],"change_id":"I7ca0ab9e42352645e463c49f8ce58793115543c0","subject":"Allow configuration of limit of batch size for exirer","status":"MERGED","created":"2020-09-23 09:19:15.000000000","updated":"2020-09-25 14:31:41.000000000","submitted":"2020-09-25 14:30:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"753550-1601044216638-e4987785","meta_rev_id":"4d04b14ffda1d4626582b660410836f8c8ac5c37","_number":753550,"virtual_id_number":753550,"owner":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-25 14:30:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-25 12:46:41.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-25 09:16:33.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-25 12:46:41.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-23 17:00:43.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-09-25 09:16:33.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-09-25 12:46:41.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-25 14:30:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a8d6fed31e3999773b224c895cd461dddce499c2","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-09-23 09:19:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0d26468385135359652aa002b2ca0c3dcb92a26","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-09-23 09:28:11.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"71c70d39845ad0180781b4e299d07f0618821034","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 10:41:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/613b75cd2656467194777b7c88bc948e : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3db8a2ec41344c1c98168d50719c5f6e : SUCCESS in 4m 06s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/242287cada5c44f8a5946c2c06907433 : SUCCESS in 9m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10c94a24f1d5457ea11680ac4905dc7f : SUCCESS in 7m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/913facd231c9477696ee7f7c53a83540 : SUCCESS in 7m 41s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6ab5ca569c5b466c8c7befa7aa4dfdc5 : SUCCESS in 48m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5aff60c34da24d3383d9453bb8df5bee : SUCCESS in 1h 04m 11s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/721601b1436a42689293fe7ffbcc7c12 : SUCCESS in 18m 08s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fd6c7bd2ac644fc927c5a7c16741f8a : FAILURE in 12m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cfe652479bb13c391436a1614b0ccba8e8671f98","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-23 13:49:01.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nThe proposed change looks good in its logic level, but needs some consideration about the parameter name and the default value.\n\nAlso, could you please add the following 2 items ?\n - unit test cases to cover the new parameter\n - a release note to explain the new parameter\n\nFeel free to let me know if you need any assistance.","accounts_in_message":[],"_revision_number":2},{"id":"0db7fd7f722b1fe0bbbd086e51ad743b9dd95696","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-09-23 16:48:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e4682c4f021705685bd2edf60f77dbde482005be","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-23 17:00:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e4e892f90d912df824d0b4596d4b42cbe3199b5d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-23 17:01:19.000000000","message":"Patch Set 3:\n\ns/OS SERVICE DEFAULT/SERVICE DEFAULT/g","accounts_in_message":[],"_revision_number":3},{"id":"283db5ed78b5325e95d551ae835432aa3bb69d8d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-23 17:04:23.000000000","message":"Patch Set 3:\n\nThe storage options seems very scattered, one is in panko::api, one is added to panko::expirer in this patch hm","accounts_in_message":[],"_revision_number":3},{"id":"da4f0661c2c18eb71b52c8ccb6e05b0245657d73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-23 18:53:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b7c5e99f9c74802af205584f5c933c9 : FAILURE in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f65f8a29c6fc4775af9b9c99c1e15725 : SUCCESS in 5m 06s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/374851f96b6e45e48de3a3b0709619d5 : SUCCESS in 12m 58s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/244f21aed3784c84a85549a7df0233f9 : SUCCESS in 9m 28s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72ef0b13ffa449a581421148e7eccaca : SUCCESS in 9m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/b0137f9b8ec14012be96087fe40ec597 : SUCCESS in 47m 25s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3a707e0951314727b81bbb91d1319b39 : SUCCESS in 1h 42m 55s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f2760f897286463bb25eb129deaf1c15 : SUCCESS in 21m 23s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7cf1cc5b53c47df94d73b44718ab627 : FAILURE in 15m 36s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e296e129071a431396d2d9f3a4562b8b : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":3},{"id":"885c51fc7fbe7c7bf6e5b1ce12c430a19207c5ca","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-23 23:12:40.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ca69780a352ae4c6746f23d038a42768b6a7760a","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-09-24 08:52:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3c476f4449001417875e1c11c36ca2772d17446f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 10:32:39.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a693b370ff32479ea35f87f5183f2b9c : SUCCESS in 4m 32s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a69fd3c079894854b6c1144ee8e8d903 : SUCCESS in 5m 30s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e2d9f4225da346708cb95fbfe832e234 : FAILURE in 10m 16s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddbfa21e5aac4447a99e2385c04e3f2f : FAILURE in 7m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77d7814a520d4f70997b5eef315a74eb : FAILURE in 8m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/32a33b741dc243308c54a95cb7be7c95 : SUCCESS in 48m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2f8705b4f80b44beb0639503620b1090 : SUCCESS in 1h 20m 19s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/4c48f261f0c042dabf70767af5022af9 : SUCCESS in 17m 02s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b80ef2f7d268457e880b1b2681747fb6 : FAILURE in 12m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c9dbe0bd4b247419c21f8a21ed92070 : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":4},{"id":"5fca173d88a695897af6e3b8db3c594717eb5fdf","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-09-24 11:03:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ab9f0b3d40ca2b1364d259977d157e210ffd1e21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 12:25:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad5b4eceaf0c4ad4ab190edc5cf1a193 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fed69bae98f1469c9f04c9693a30d728 : SUCCESS in 4m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b090a80b76834e6989781255c958109a : FAILURE in 10m 02s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d41692acba743dba1f776c9126bca57 : FAILURE in 8m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5e50d16ef694dbaa3cd4fed6c459188 : FAILURE in 9m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/97bdfd66c38349e79249e943ddd4f94f : SUCCESS in 52m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bc1726aaa3234abba89b16edda6ef0c0 : SUCCESS in 1h 15m 55s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/9feec4b953c448c986f536e728d70058 : SUCCESS in 20m 47s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/443338deeda341c29fa8f104ef2e57a8 : FAILURE in 17m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffd8a7a06f1f466a9ea720a2d26f0656 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":5},{"id":"b2648a6aa33e833be248e7b1c9149eca55d6c949","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-09-24 12:36:56.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"983890083c1f97a315e110ffc6167821b32a3c0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-24 14:06:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47f4a6c6068d4b78a6cc751b38ce9da6 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d080b4868ef4b98a37887c44d59c515 : SUCCESS in 3m 59s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2bad34d40bdd450881bae572e34227a8 : SUCCESS in 10m 20s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0bfb1a72f139452ea36c181a6f550e6d : SUCCESS in 7m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a7d74e13655348a48116014462cea3b4 : SUCCESS in 7m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/3af3aa0c16e544dcb5b593032ceb5eff : SUCCESS in 48m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cbb76ea487274670a6594e52243f88ce : SUCCESS in 1h 21m 41s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/7d57e1c152d9406a8e42b17da3495bdb : SUCCESS in 18m 46s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8a7396d4fce46aebcb3a4fecfb9c92a : FAILURE in 15m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fef3d91280664edca2d6d948cc848347 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":6},{"id":"29b5734aaa1c672417f439dc97631e00853f9d60","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-25 01:04:16.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"57507848d5d7fd11f6a1a4822b989d83f43126af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 02:05:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ac7753129d24ea8834ce6f7a4c21d90 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ced5923333474046b10224b235bfa341 : SUCCESS in 5m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/693fb1998080448f8587307d5e4bf53e : SUCCESS in 8m 19s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e4c432efa6e4c038869c27ff0abcf2f : SUCCESS in 9m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4752996b6ccf4bb9b095335cb5062651 : SUCCESS in 7m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/6ab947896120463492fd850a31944caa : SUCCESS in 52m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b9f095c205f14cd88c67fdb27b70d5d9 : SUCCESS in 58m 23s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/62b4f60c1ba642618e83910f0c153eb6 : SUCCESS in 18m 06s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d5c115030104e0888e1fde0403adae2 : FAILURE in 29m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdf5f6ae997d491bad320019f2c109ef : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":7},{"id":"61c07729cae8cee6ebb48372854a1e91bf49721a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-09-25 08:50:14.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"08173fa31c66670f8199bd752dc162420975c503","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-25 09:16:33.000000000","message":"Patch Set 8: Code-Review+2\n\nThanks for fixing unnecessary space!","accounts_in_message":[],"_revision_number":8},{"id":"c002fff038f929689023df70343c60025faaf166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 10:08:36.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f56ed684d4354a75843796d0d8bca414 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b41992c43064361885679dfffab7db4 : SUCCESS in 3m 08s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0c2b7011c36641928e2967a3ff20931b : SUCCESS in 13m 59s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6111f1a142cc4e25ba65feefca753f05 : SUCCESS in 12m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/266d82db87294015bac0ecefdef4f552 : SUCCESS in 10m 31s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/025ab133851046c5bca9fedd7999a882 : SUCCESS in 53m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f743259db48747e1a0f49773f265176f : SUCCESS in 1h 11m 11s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f61c33b3df2c4e0cad1977f337b06b44 : SUCCESS in 20m 41s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58c430202c3a4220898ba5b31485f439 : FAILURE in 14m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b787ba6e29448da941bc22c205e9718 : POST_FAILURE in 6m 20s","accounts_in_message":[],"_revision_number":8},{"id":"f7af7f3e79a95086fb8ef8257d778c46f757cef1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-25 11:29:12.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"abae5a9c70d5eee8d4bc167e43820b67b1322217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 12:45:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d6bc07f101a447d89ed40c0f031c4cca : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b4e28abad374df4958794942ca7394a : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/be485b943277436cb6742ff47449a125 : SUCCESS in 10m 28s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/85596abf5e974d348b41861742b9120a : SUCCESS in 9m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e7e2499494d49388a485331a5d18c9b : SUCCESS in 9m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/346cc404125c4fdc94698616b99601b3 : SUCCESS in 54m 29s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/febfee20683e45c4b8257caa6efe82cb : SUCCESS in 1h 13m 03s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/914a21563ad847d6a79de1680c0b6166 : SUCCESS in 19m 58s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e38b7fc3cedc4f18ada9785d01c551e0 : FAILURE in 17m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19724c70d21a467892133c33c0d89741 : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":8},{"id":"9dbd560f32cda2829d4f9bc75d0cd4cecf05c9b1","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-25 12:46:41.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"ed3bf7d5860fdddd722dddfa38596f9fb1fd0c14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 12:46:53.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"620516caf227af2895111447d49a9ade76a97a3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 14:30:16.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9454853f4704b3589746b7cda7f5c52 : SUCCESS in 3m 19s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0626046aab54c1bac16ee90d18a8439 : SUCCESS in 5m 32s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/27d6c1bef1fe4ba7bea23f43c5b4d3b0 : SUCCESS in 11m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccf0cbeabdea424a91933b8c8c7c7139 : SUCCESS in 7m 11s\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/070e6abcb1dd4c949dfb443032d26ddd : SUCCESS in 52m 24s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fadeab872c544adaa69fa400bcaf4545 : SUCCESS in 1h 35m 50s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d78db0cd225e491692dc6dbe4f01a110 : SUCCESS in 19m 10s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb86658beedd47548ba5b351dcc8fb55 : FAILURE in 15m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01b7049434044520bb4fd11d43269e14 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":8},{"id":"3b47cff1d0ae13bc15ce770c4784edd9ebdce91a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 14:30:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"4d04b14ffda1d4626582b660410836f8c8ac5c37","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-25 14:31:41.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da3bb0ec8df944bb9b5881cdce274211 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"89c3f384f01b778aee6a2c7ce24db226b0fbd770","revisions":{"1896d5cde2c9e940bb071db7981dd45a9df014a6":{"kind":"REWORK","_number":1,"created":"2020-09-23 09:19:15.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/50/753550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/50/753550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/1"}}},"commit":{"parents":[{"commit":"0c2d356aeac2cbeaa516a429b15b9d58d3464fe2","subject":"Set openstackdocs_auto_name to use \u0027project\u0027 as name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/0c2d356aeac2cbeaa516a429b15b9d58d3464fe2"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-23 09:17:17.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-23 09:19:13.000000000","tz":120},"subject":"Allow configuration of limit of batch size for exirer","message":"Allow configuration of limit of batch size for exirer\n\nAdds new configuration option for panko-expirer for limiting detele\nbatch size.\n\nDepends-On: Icf83ffe089301b3782273923f18efd4d209131c2\nChange-Id: I7ca0ab9e42352645e463c49f8ce58793115543c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/1896d5cde2c9e940bb071db7981dd45a9df014a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/1896d5cde2c9e940bb071db7981dd45a9df014a6"}]},"branch":"refs/heads/master"},"1ffbc29ab2ed4c53b71599d4596d7e19d95e3b08":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-09-23 09:28:11.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/50/753550/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/50/753550/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/2"}}},"commit":{"parents":[{"commit":"0c2d356aeac2cbeaa516a429b15b9d58d3464fe2","subject":"Set openstackdocs_auto_name to use \u0027project\u0027 as name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/0c2d356aeac2cbeaa516a429b15b9d58d3464fe2"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-23 09:17:17.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-23 09:27:50.000000000","tz":120},"subject":"Allow configuration of limit of batch size for exirer","message":"Allow configuration of limit of batch size for exirer\n\nAdds new configuration option for panko-expirer for limiting detele\nbatch size.\n\nChange-Id: I7ca0ab9e42352645e463c49f8ce58793115543c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/1ffbc29ab2ed4c53b71599d4596d7e19d95e3b08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/1ffbc29ab2ed4c53b71599d4596d7e19d95e3b08"}]},"branch":"refs/heads/master"},"9636a23ffd6b6377939cba79ef5b3d43d2e13ee0":{"kind":"REWORK","_number":3,"created":"2020-09-23 16:48:20.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/50/753550/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/50/753550/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/3"}}},"commit":{"parents":[{"commit":"0c2d356aeac2cbeaa516a429b15b9d58d3464fe2","subject":"Set openstackdocs_auto_name to use \u0027project\u0027 as name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/0c2d356aeac2cbeaa516a429b15b9d58d3464fe2"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-23 09:17:17.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-23 16:48:10.000000000","tz":120},"subject":"Allow configuration of limit of batch size for exirer","message":"Allow configuration of limit of batch size for exirer\n\nAdds new configuration option for panko-expirer for limiting detele\nbatch size.\n\nChange-Id: I7ca0ab9e42352645e463c49f8ce58793115543c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/9636a23ffd6b6377939cba79ef5b3d43d2e13ee0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/9636a23ffd6b6377939cba79ef5b3d43d2e13ee0"}]},"branch":"refs/heads/master"},"b8e6ee12b546f752088a69d54eb1920e1b95fe04":{"kind":"REWORK","_number":4,"created":"2020-09-24 08:52:01.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/50/753550/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/50/753550/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/4"}}},"commit":{"parents":[{"commit":"0c2d356aeac2cbeaa516a429b15b9d58d3464fe2","subject":"Set openstackdocs_auto_name to use \u0027project\u0027 as name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/0c2d356aeac2cbeaa516a429b15b9d58d3464fe2"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-23 09:17:17.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-24 08:51:50.000000000","tz":120},"subject":"Allow configuration of limit of batch size for exirer","message":"Allow configuration of limit of batch size for exirer\n\nAdds new configuration option for panko-expirer for limiting detele\nbatch size.\n\nChange-Id: I7ca0ab9e42352645e463c49f8ce58793115543c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/b8e6ee12b546f752088a69d54eb1920e1b95fe04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/b8e6ee12b546f752088a69d54eb1920e1b95fe04"}]},"branch":"refs/heads/master"},"6f8743d810580937bdab34e4b1ab6ca377ddf0e2":{"kind":"REWORK","_number":5,"created":"2020-09-24 11:03:12.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/50/753550/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/50/753550/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/5"}}},"commit":{"parents":[{"commit":"0c2d356aeac2cbeaa516a429b15b9d58d3464fe2","subject":"Set openstackdocs_auto_name to use \u0027project\u0027 as name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/0c2d356aeac2cbeaa516a429b15b9d58d3464fe2"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-23 09:17:17.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-24 11:03:03.000000000","tz":120},"subject":"Allow configuration of limit of batch size for exirer","message":"Allow configuration of limit of batch size for exirer\n\nAdds new configuration option for panko-expirer for limiting detele\nbatch size.\n\nChange-Id: I7ca0ab9e42352645e463c49f8ce58793115543c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/6f8743d810580937bdab34e4b1ab6ca377ddf0e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/6f8743d810580937bdab34e4b1ab6ca377ddf0e2"}]},"branch":"refs/heads/master"},"7f1d614c2a05c72986734e89e51ad92f0c72bf6c":{"kind":"REWORK","_number":6,"created":"2020-09-24 12:36:56.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/50/753550/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/50/753550/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/6"}}},"commit":{"parents":[{"commit":"0c2d356aeac2cbeaa516a429b15b9d58d3464fe2","subject":"Set openstackdocs_auto_name to use \u0027project\u0027 as name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/0c2d356aeac2cbeaa516a429b15b9d58d3464fe2"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-23 09:17:17.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-24 12:36:43.000000000","tz":120},"subject":"Allow configuration of limit of batch size for exirer","message":"Allow configuration of limit of batch size for exirer\n\nAdds new configuration option for panko-expirer for limiting detele\nbatch size.\n\nChange-Id: I7ca0ab9e42352645e463c49f8ce58793115543c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/7f1d614c2a05c72986734e89e51ad92f0c72bf6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/7f1d614c2a05c72986734e89e51ad92f0c72bf6c"}]},"branch":"refs/heads/master"},"0e196e7c0d8afc522dc7c335c929d9ac752713f3":{"kind":"REWORK","_number":7,"created":"2020-09-25 01:04:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/50/753550/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/50/753550/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/7"}}},"commit":{"parents":[{"commit":"0c2d356aeac2cbeaa516a429b15b9d58d3464fe2","subject":"Set openstackdocs_auto_name to use \u0027project\u0027 as name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/0c2d356aeac2cbeaa516a429b15b9d58d3464fe2"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-23 09:17:17.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-25 01:04:07.000000000","tz":0},"subject":"Allow configuration of limit of batch size for exirer","message":"Allow configuration of limit of batch size for exirer\n\nAdds new configuration option for panko-expirer for limiting detele\nbatch size.\n\nChange-Id: I7ca0ab9e42352645e463c49f8ce58793115543c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/0e196e7c0d8afc522dc7c335c929d9ac752713f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/0e196e7c0d8afc522dc7c335c929d9ac752713f3"}]},"branch":"refs/heads/master"},"89c3f384f01b778aee6a2c7ce24db226b0fbd770":{"kind":"REWORK","_number":8,"created":"2020-09-25 08:50:14.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/50/753550/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/50/753550/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/50/753550/8"}}},"commit":{"parents":[{"commit":"0c2d356aeac2cbeaa516a429b15b9d58d3464fe2","subject":"Set openstackdocs_auto_name to use \u0027project\u0027 as name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/0c2d356aeac2cbeaa516a429b15b9d58d3464fe2"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-23 09:17:17.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2020-09-25 08:50:08.000000000","tz":0},"subject":"Allow configuration of limit of batch size for exirer","message":"Allow configuration of limit of batch size for exirer\n\nAdds new configuration option for panko-expirer for limiting detele\nbatch size.\n\nChange-Id: I7ca0ab9e42352645e463c49f8ce58793115543c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/89c3f384f01b778aee6a2c7ce24db226b0fbd770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/89c3f384f01b778aee6a2c7ce24db226b0fbd770"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
