)]}'
{"id":"openstack%2Fpuppet-panko~755265","triplet_id":"openstack%2Fpuppet-panko~stable%2Fussuri~I7ca0ab9e42352645e463c49f8ce58793115543c0","project":"openstack/puppet-panko","branch":"stable/ussuri","hashtags":[],"change_id":"I7ca0ab9e42352645e463c49f8ce58793115543c0","subject":"Allow configuration of limit of batch size for exirer","status":"MERGED","created":"2020-09-30 12:01:19.000000000","updated":"2020-10-01 12:45:01.000000000","submitted":"2020-10-01 12:43:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"755265-1601556222752-ef4f6a00","meta_rev_id":"6201a6490837b78d3e973f12f9c61c13db9fdc1c","_number":755265,"virtual_id_number":755265,"owner":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-01 12:43:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-30 14:33:02.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-10-01 11:44:22.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-10-01 11:44:22.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-30 14:33:02.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-01 11:44:22.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-10-01 12:43:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"827999281fc68b0bd1019dd11165468ee865cb31","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-09-30 12:01:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a0128855cc2ddebe581d5afb137fa1e25f97b88","author":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"date":"2020-09-30 12:02:15.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 8fcdfa334270ad8f1070326ec0e85e852c186ce4","accounts_in_message":[],"_revision_number":1},{"id":"1f833718d381e33739ebb4d5927b7ea639ec7d91","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-30 13:03:34.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"7380dc2743e58230b7908f39ce332f85fdf52500","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-09-30 13:06:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f71510adaebf3ffbf82361b1091393a073c143d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-30 14:06:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f65f5d8c6934c41a27ec04bc6d92676 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1be65ae51b6445391414e802dc4136f : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e90b0f7ed8f4b8a8be2cbec326b1f68 : SUCCESS in 4m 30s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8f2562ee4b4041489e787970695d4be0 : SUCCESS in 15m 50s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/c4a3858835794970b4b89257a178ebdc : SUCCESS in 13m 22s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/46003ca8b56c41ffa783886a866b252b : SUCCESS in 12m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3191e73c2e8b4e8d8a7f7f268022705c : SUCCESS in 9m 58s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/9ae082ec92574fa997854df0162ba303 : SUCCESS in 48m 12s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/28c8d3a61194413d8c8a2371b465f5be : SUCCESS in 56m 01s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/2c8158a0ed0c4cc59a1f8e8c568147fd : FAILURE in 8m 20s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d686bfbc9f0147f389f14634c40bf129 : FAILURE in 7m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcca5a5a771e4a6fafe92a9cec15af77 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":2},{"id":"2820fe894b086fd207414a34823e9d470db923fb","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-30 14:33:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"463d44724474f3f0c4077ea807b2abf8cc000d95","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-01 11:44:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aa71d79d7c635465381c17ec43171700a573d7c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 11:44:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2db1bdce4acd003f71b901fd85c8b7e8067cf287","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 12:43:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ec49a43888a511e29bc15a26b8257a9c05b6436b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 12:43:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/173260c928ad4792a799c40a17395920 : SUCCESS in 4m 27s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9198789505f469e840bb5ef6ad8c438 : SUCCESS in 4m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d813f2d4ab524f559960ce284265ee7b : SUCCESS in 4m 38s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b45f193c184b4947b8252e539a71657e : SUCCESS in 10m 17s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/efafb8e7aecc4b2982f3fcdc13a661c0 : SUCCESS in 11m 28s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6aa812363f7642dbb424487e4520fd46 : SUCCESS in 7m 40s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7ecae415515f482ea800b62db78b7f26 : SUCCESS in 44m 12s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/47ca2bbcb8494452a0ac92b86ecb6a63 : SUCCESS in 53m 05s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9f314b7c278e48e6afeab44263791140 : FAILURE in 9m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc607298d6b54b8aac7616b5f6b5d5e8 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":2},{"id":"6201a6490837b78d3e973f12f9c61c13db9fdc1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-01 12:45:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbf4794de57e43f7b057132473b59d0d : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c954ebf149edaa0ccfe884d299cb2f05aa783100","revisions":{"67967997d290de7d79670a1fabc63376712c756e":{"kind":"REWORK","_number":1,"created":"2020-09-30 12:01:19.000000000","uploader":{"_account_id":5241,"name":"Martin Magr","email":"mmagr@redhat.com","username":"mmagr"},"ref":"refs/changes/65/755265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/65/755265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/65/755265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/65/755265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/65/755265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/65/755265/1"}}},"commit":{"parents":[{"commit":"4712c01716f9e39eb38ca51082480c7bb1699625","subject":"Add CentOS8 to nodesets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/4712c01716f9e39eb38ca51082480c7bb1699625"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-23 09:17:17.000000000","tz":120},"committer":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-30 12:00:48.000000000","tz":120},"subject":"Allow configuration of limit of batch size for exirer","message":"Allow configuration of limit of batch size for exirer\n\nAdds new configuration option for panko-expirer for limiting detele\nbatch size.\n\nChange-Id: I7ca0ab9e42352645e463c49f8ce58793115543c0\n(cherry picked from commit 89c3f384f01b778aee6a2c7ce24db226b0fbd770)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/67967997d290de7d79670a1fabc63376712c756e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/67967997d290de7d79670a1fabc63376712c756e"}]},"branch":"refs/heads/stable/ussuri"},"c954ebf149edaa0ccfe884d299cb2f05aa783100":{"kind":"REWORK","_number":2,"created":"2020-09-30 13:03:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/65/755265/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/65/755265/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/65/755265/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/65/755265/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/65/755265/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/65/755265/2"}}},"commit":{"parents":[{"commit":"4712c01716f9e39eb38ca51082480c7bb1699625","subject":"Add CentOS8 to nodesets","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/4712c01716f9e39eb38ca51082480c7bb1699625"}]}],"author":{"name":"Martin Magr","email":"mmagr@redhat.com","date":"2020-09-23 09:17:17.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-09-30 13:03:29.000000000","tz":0},"subject":"Allow configuration of limit of batch size for exirer","message":"Allow configuration of limit of batch size for exirer\n\nAdds new configuration option for panko-expirer for limiting detele\nbatch size.\n\nChange-Id: I7ca0ab9e42352645e463c49f8ce58793115543c0\n(cherry picked from commit 89c3f384f01b778aee6a2c7ce24db226b0fbd770)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/c954ebf149edaa0ccfe884d299cb2f05aa783100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/c954ebf149edaa0ccfe884d299cb2f05aa783100"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
