)]}'
{"id":"openstack%2Fpuppet-panko~756953","triplet_id":"openstack%2Fpuppet-panko~master~I3bd31e64eac9ee98bbf7863b52107e6942f486dc","project":"openstack/puppet-panko","branch":"master","topic":"missing-deps","hashtags":[],"change_id":"I3bd31e64eac9ee98bbf7863b52107e6942f486dc","subject":"Include deps class in unit test for mysql","status":"MERGED","created":"2020-10-09 02:00:41.000000000","updated":"2020-10-09 03:23:04.000000000","submitted":"2020-10-09 03:23:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756953-1602213784242-5cec80e2","meta_rev_id":"6f40b612c141034bb5ec0e8874c591b209be8cf9","_number":756953,"virtual_id_number":756953,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-09 03:23:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-10-09 03:06:06.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-10-09 03:06:06.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-09 03:23:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ece47da5511b3e65573d505e30f377db4c812571","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-09 02:00:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f1f59bacfae7e0556d651ab56caec4cfe61dc9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 02:17:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dfac2c7d57b040aa830c5f3cf2b3d191 : SUCCESS in 2m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/399293b4314e4b27ad3f0ec464f86db7 : SUCCESS in 3m 04s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ce2ff8e6af6f465b98b05881b0349a6c : FAILURE in 9m 28s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42f04c425ff7409298348002c46083e2 : FAILURE in 7m 59s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a03642714cb447ef9315684c410420d7 : FAILURE in 9m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7cc7bb0e7361a89caacddec7bb4c2175bbf8991a","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-09 02:33:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5934065dffa59ddd1d8bde2b5a3a33f387dd333","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 02:49:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae69e98ab0fd4cd1acfbf27364a49416 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19a6ff5857df49a4bbc199247ee67b94 : SUCCESS in 3m 05s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/681dc98f8f17418dbcf37f61da9cdf33 : SUCCESS in 10m 36s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5336c8c6ec7f41c3b983845b16626b2a : SUCCESS in 8m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e11ec45528d842be9fc84e1563202458 : SUCCESS in 8m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e66e30762e2e86c700a5696fef9c73ebeea67442","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-09 03:06:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a7d4ed3094fecd358e29b07e349a147d5bb980fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:06:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d85ac207f6e1996be18d5906d0dd412dffe285dc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:23:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"594701835470c39d548064f7f38e1c09cade6eb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-09 03:23:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82d889ce1ffb4f2c8ea4a2985cb0e73d : SUCCESS in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d62b21c476544ff98da32c0e7b6082fd : SUCCESS in 3m 10s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/140f2014928941febc47d579d01a09c7 : SUCCESS in 9m 30s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/14498d613ca845cd85a0e9e01a0e687e : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"89c2b417169863a191249a5ab56c38ae4ce893bd","revisions":{"69808a63978e9f4b92f5f4eec81635837cfa9f11":{"kind":"REWORK","_number":1,"created":"2020-10-09 02:00:41.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/53/756953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/53/756953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/53/756953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/53/756953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/53/756953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/53/756953/1"}}},"commit":{"parents":[{"commit":"7c0990900acb03b4c7428f38ea6219b2447e8969","subject":"Prepare a new release for victoria cut off","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/7c0990900acb03b4c7428f38ea6219b2447e8969"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-09 01:59:55.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-09 01:59:55.000000000","tz":480},"subject":"Include deps class in unit test for mysql","message":"Include deps class in unit test for mysql\n\nChange-Id: I3bd31e64eac9ee98bbf7863b52107e6942f486dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/69808a63978e9f4b92f5f4eec81635837cfa9f11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/69808a63978e9f4b92f5f4eec81635837cfa9f11"}]},"branch":"refs/heads/master"},"89c2b417169863a191249a5ab56c38ae4ce893bd":{"kind":"REWORK","_number":2,"created":"2020-10-09 02:33:12.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/53/756953/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/53/756953/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/53/756953/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/53/756953/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/53/756953/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/53/756953/2"}}},"commit":{"parents":[{"commit":"7c0990900acb03b4c7428f38ea6219b2447e8969","subject":"Prepare a new release for victoria cut off","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/7c0990900acb03b4c7428f38ea6219b2447e8969"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-09 01:59:55.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-10-09 02:33:02.000000000","tz":480},"subject":"Include deps class in unit test for mysql","message":"Include deps class in unit test for mysql\n\nChange-Id: I3bd31e64eac9ee98bbf7863b52107e6942f486dc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/89c2b417169863a191249a5ab56c38ae4ce893bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/89c2b417169863a191249a5ab56c38ae4ce893bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
