)]}'
{"id":"openstack%2Fpuppet-panko~784503","triplet_id":"openstack%2Fpuppet-panko~master~Ibfc588495de4f0d65bd7a1900fae870ada2d50c0","project":"openstack/puppet-panko","branch":"master","topic":"policy-dirs","hashtags":[],"change_id":"Ibfc588495de4f0d65bd7a1900fae870ada2d50c0","subject":"Allow to configure policy_dirs","status":"MERGED","created":"2021-04-01 21:24:11.000000000","updated":"2021-04-06 08:48:28.000000000","submitted":"2021-04-06 08:47:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784503-policy-dirs","meta_rev_id":"4945f35d3dba830c3f8a7476a59656cf94ece537","_number":784503,"virtual_id_number":784503,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-06 08:47:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-04-05 21:29:41.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-04-06 07:19:41.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-04-06 07:19:41.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 23:08:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-05 21:29:41.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-04-06 07:19:41.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"70575d42d918e751f3d729f21bbe0c6ab7ee5c2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-01 21:24:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3caa742ae2bb44153e6c1ab19a799e62b9179724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 23:08:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc562a83f13f4d8280241a8e7acc134d : SUCCESS in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd1c1222a8b94308830e735a7a70473f : SUCCESS in 3m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9b3a30d5991e40d5b93939ef131d87ad : SUCCESS in 11m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8618b7d1dda4bbab1f1b11ea04b5ec3 : SUCCESS in 6m 54s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cb2447458b94cefbef7fc74b82dc23f : SUCCESS in 6m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8cb85f9a5a346f28a68286ee3c08329 : SUCCESS in 58m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6292e9cdff244716b3f5995da632f656 : SUCCESS in 1h 36m 17s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/ae25ff1cc78749709372de68139111e4 : SUCCESS in 18m 36s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f109fbf233094d71afef5e46439d5c96 : FAILURE in 12m 35s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c75381fda7240449c51825ac694877e : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":1},{"id":"b1843e601ca4d8ec1b298a2effa9cd38e2a28c07","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-05 21:29:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9354e52caa27d027e92ca2dec56bf62fcc963385","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-04-06 07:19:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87753bf461ce9fbc56f276f68425954f8622ce39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 07:20:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"34f320bb3fd65b89eb3d74ff43293e58d8468d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 08:47:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f05960211f434401a51cdaa34ab43742 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/243badc82e91402e8f2fa0f80a043cf8 : SUCCESS in 3m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f88c8f13ad94dca8e2ed79e9ffc08c4 : SUCCESS in 9m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16918277547c428bbef9577d591f9ead : SUCCESS in 6m 57s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/decb86125b4e48b0b1c6e3f82fdb1ea6 : SUCCESS in 1h 21m 45s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/2bd56a4ff8f5402fa5affc45b09c42ea : SUCCESS in 19m 50s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b8ac3ad50c0470f898f2781a3ba2ada : FAILURE in 13m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/723cc43904b84112b3d7a5996b2b8522 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"814e17f2e6b40b45f92b5059b97051f20a78e0a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 08:47:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4945f35d3dba830c3f8a7476a59656cf94ece537","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 08:48:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/098ee067b9ce47759d153199a49ce3f0 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"df2a48bfbcc712f002836f01c7e1c7ea5d62b11d","revisions":{"df2a48bfbcc712f002836f01c7e1c7ea5d62b11d":{"kind":"REWORK","_number":1,"created":"2021-04-01 21:24:11.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/03/784503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/03/784503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/03/784503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/03/784503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/03/784503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/03/784503/1"}}},"commit":{"parents":[{"commit":"c83a1d5b37cda0320470d6c9297dc1da5c028737","subject":"Merge \"metadata.json: debian 11\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/c83a1d5b37cda0320470d6c9297dc1da5c028737"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 21:23:58.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-01 21:23:58.000000000","tz":120},"subject":"Allow to configure policy_dirs","message":"Allow to configure policy_dirs\n\nThis patch makes it possible to override the current service default,\nwhich is /etc/\u003cservice\u003e/policy.d .\n\nChange-Id: Ibfc588495de4f0d65bd7a1900fae870ada2d50c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/df2a48bfbcc712f002836f01c7e1c7ea5d62b11d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/df2a48bfbcc712f002836f01c7e1c7ea5d62b11d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
