)]}'
{"id":"openstack%2Fpuppet-panko~787257","triplet_id":"openstack%2Fpuppet-panko~master~I6c805de4681f2aa0832d9bdd7b636f2dcf479256","project":"openstack/puppet-panko","branch":"master","topic":"hc-typo","hashtags":[],"change_id":"I6c805de4681f2aa0832d9bdd7b636f2dcf479256","subject":"Fix a typo in file header","status":"MERGED","created":"2021-04-20 16:10:45.000000000","updated":"2021-04-27 13:20:33.000000000","submitted":"2021-04-27 13:20:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787257-hc-typo","meta_rev_id":"0eb56173a2013973c406969005aac0231ab6567b","_number":787257,"virtual_id_number":787257,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-27 13:20:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-04-22 15:21:05.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-04-27 08:55:48.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-04-27 08:55:48.000000000","_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 17:47:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 15:21:05.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-04-27 08:55:48.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"c414def2f8387020947a3b644f67560345644e19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-20 16:10:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24e6cf15dba205e099a7682008dcda3650ed47fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 17:47:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1dffe7a26073487996d6923d0060f476 : SUCCESS in 4m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fd3f48cd57d43f8af51ac4c05faa489 : SUCCESS in 5m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8073132e279545d7bd65a40ac56de6dd : SUCCESS in 11m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/816bb6218eab46ec8e0d060f1a8e019f : SUCCESS in 7m 20s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e601e642b7642f79ffedb6e499c17c0 : SUCCESS in 7m 05s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be8a56f825664e8f869b6e7ccc5392cb : FAILURE in 30m 49s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5bef0ca7689545348a22a2b6b3ee2c0d : SUCCESS in 1h 30m 30s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/05593bef0e07497b8ad5519c8f197461 : SUCCESS in 22m 22s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa5dfb65e9cd48eaaa7c3859aa6af4bc : FAILURE in 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f91862d8fae259e80df38e2163e506271ca3e442","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-22 15:21:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"67b06b54e8963f9116064551bf4d0bdd7feefeb4","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-04-27 08:55:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"39a6d61f75f5c44bea6e0985261e877eeeb78e27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 08:58:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a90ebfe36aae3b88a530508fa63a8cfa580e2a86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 10:50:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0febe9df4858438c8e606e15191b4559 : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94cfdacd7b804da4a7677f5026f2c34e : SUCCESS in 3m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f211ffee2aae4e3bb476655a214ed5c7 : SUCCESS in 9m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3dca55dbfc748ca97e8bc5246a13962 : SUCCESS in 7m 49s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3793dd09556846a7b78ef348d272d75c : FAILURE in 1h 43m 52s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/d5826330a6694582846eceac716683f8 : SUCCESS in 19m 31s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0033c72689ec4be4ba6ad6a6c54b1d39 : FAILURE in 15m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2e586be142e2bd908ba412dd3d193b4eecd1478","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-27 10:55:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6afeed3b82c5a9ab41f4035c2108509f23346c48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 12:06:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e84f8d68c3f4e89af2d53d105c5d2cd : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/006e8fe0605b470ba5e5dfce15653d48 : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/790c5d7cfe614a27ab57b0c341220fe9 : SUCCESS in 11m 14s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/788909fc1fa247bcad77e65476bc2bde : SUCCESS in 7m 58s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/064815557b884597924b1fada45d63e9 : SUCCESS in 7m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d792a85358444f068313ce694161b333 : SUCCESS in 56m 49s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/87517c35355d4f20bf236f30a67512f3 : SUCCESS in 1h 05m 52s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/0134b9d2754049e6ae002fd4f65e2649 : SUCCESS in 18m 39s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a5056a38d2c419582b1004c42b9ae5b : FAILURE in 18m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2a56eae0b165c5fc8c60fdaee7d5ba61ebcf1859","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 12:08:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e17be973e18b8fca9d540c04799baced553412bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 13:20:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6eec669c91214a09a7ec08db944c1c80 : SUCCESS in 4m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79e857c83d024ff4a47268342c916c4f : SUCCESS in 3m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/69430125aba848ed8c5994d7d816806b : SUCCESS in 11m 28s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc5973faf5144bf29c0e0032046512f7 : SUCCESS in 9m 22s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec0bc77f79f146f8abf7a02ac1d33ca5 : SUCCESS in 1h 08m 46s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8c78615298940dca027669bf37f2246 : SUCCESS in 19m 00s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ec680dab3a74ac5aa98ef882a6abcd1 : FAILURE in 14m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0eb56173a2013973c406969005aac0231ab6567b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 13:20:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8fbf6aa52ba2d5346e12517bd0c504b1b053b3cb","revisions":{"8fbf6aa52ba2d5346e12517bd0c504b1b053b3cb":{"kind":"REWORK","_number":1,"created":"2021-04-20 16:10:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/57/787257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/57/787257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/57/787257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/57/787257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/57/787257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/57/787257/1"}}},"commit":{"parents":[{"commit":"d0ce60cc3616f4bb8ea0b6fefe1bfb5a3e904320","subject":"Merge \"Add support for panko_api_uwsgi_config in Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/d0ce60cc3616f4bb8ea0b6fefe1bfb5a3e904320"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-20 16:10:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-20 16:10:36.000000000","tz":540},"subject":"Fix a typo in file header","message":"Fix a typo in file header\n\nChange-Id: I6c805de4681f2aa0832d9bdd7b636f2dcf479256\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/8fbf6aa52ba2d5346e12517bd0c504b1b053b3cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/8fbf6aa52ba2d5346e12517bd0c504b1b053b3cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
