)]}'
{"id":"openstack%2Fpuppet-panko~787272","triplet_id":"openstack%2Fpuppet-panko~master~Iadf4f00adbb35b832fce7e539e8d76aaecab9acd","project":"openstack/puppet-panko","branch":"master","topic":"init-settings-no-openstack-settings","hashtags":[],"change_id":"Iadf4f00adbb35b832fce7e539e8d76aaecab9acd","subject":"Debian api_uwsgi_config: Switch to :ini_setting","status":"MERGED","created":"2021-04-20 17:20:26.000000000","updated":"2021-04-26 14:54:55.000000000","submitted":"2021-04-26 14:54:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787272-init-settings-no-openstack-settings","meta_rev_id":"3c434c3e69c92d9044b6d6754158751781165638","_number":787272,"virtual_id_number":787272,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 14:54:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-04-26 13:12:11.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-04-22 15:21:12.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-04-26 13:12:11.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 19:08:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-22 15:21:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-04-26 13:12:11.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"e82cc0fe2e1eabb2ac55fcfb843a38d9c07c9f8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-20 17:20:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"056d3b75683e9214dca972e712576dc70456ecee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 19:08:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0b845535a3a46999ee6d689f11b32ae : SUCCESS in 4m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/971c50d3382f4189863b6f4965b4cae4 : SUCCESS in 4m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa022033721d4d45825da90ab8a842cd : FAILURE in 9m 25s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71141f4c31ef4722b976fefa6bbdde77 : FAILURE in 6m 54s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bce457045fb24dccb1de0afcfa5e4502 : FAILURE in 11m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/209aa2c2b4864deab8d59f6370a97211 : FAILURE in 1h 21m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f46d6901782b4e499a754ea12978352b : SUCCESS in 1h 27m 43s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/410a206d7cd9471d97bc147817b43744 : SUCCESS in 18m 56s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df964359442245c68b9f844e0afdb59b : FAILURE in 17m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc19c70a940ab7d7f952bb6471e479e2f71d3c9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-21 16:13:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1f0ec11f52326e568941534a61fa0e3eab856626","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-21 16:15:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"af8d0dd022edb9dd09f4fab4ce97053a77931214","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 17:44:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84c9f81bc5d640f183a891f66c94e389 : SUCCESS in 6m 33s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/feba97d2348e46dea2474b93f5105a7e : SUCCESS in 6m 55s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d7633f1baba8470d82c10ed2de1696cd : SUCCESS in 9m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31706e90bbdf4eb3a9ef2d357468ca89 : SUCCESS in 7m 49s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de2b30490c104a6d8707ae43a79a2c49 : SUCCESS in 7m 52s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8741d885c5544bdb88357dc1668f1a56 : SUCCESS in 56m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b3ad79b1e6204410a39a04cf401d0337 : SUCCESS in 1h 22m 57s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/68e6f9519ead4f939c4c16a9fd17cd48 : SUCCESS in 17m 58s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a05b21cfc90345caab6ba7029a5ecddd : FAILURE in 23m 44s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f5f973bce556b57fa0451c0976fe6617ebfb0d83","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-22 15:21:12.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"201b2463b18215eb4c8680ff8adae0f66b1b338a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-26 13:12:11.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a14cacff790907d12e2d7f21f6fd19e25a267a2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 13:12:35.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"35609d4af7573988493d9a694272f0e99d877e6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 14:54:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3e2bcc7e9ef48e89dde229a58ca4873 : SUCCESS in 3m 50s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd49812e41b14b4291d55f1773c29513 : SUCCESS in 3m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e4504c63b23494cba15cbbc17396f4d : SUCCESS in 9m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eed5ab2a99b84ee1a3fd7c0bfc11e837 : SUCCESS in 6m 56s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/39c0794a1616466ca8a3e8f6c7ccb75e : SUCCESS in 1h 36m 38s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfa7c39b5b4f46229ac38ab31ee175bc : SUCCESS in 17m 14s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/242cab23276240178c984d02629a2f88 : FAILURE in 12m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3c434c3e69c92d9044b6d6754158751781165638","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 14:54:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"84bba8a866df95455dcd7ee7df37a692847553f4","revisions":{"754289fd9d596edc4e25e58ad9edf6399e5ea900":{"kind":"REWORK","_number":1,"created":"2021-04-20 17:20:26.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/72/787272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/72/787272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/72/787272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/72/787272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/72/787272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/72/787272/1"}}},"commit":{"parents":[{"commit":"d0ce60cc3616f4bb8ea0b6fefe1bfb5a3e904320","subject":"Merge \"Add support for panko_api_uwsgi_config in Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/d0ce60cc3616f4bb8ea0b6fefe1bfb5a3e904320"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-20 17:19:49.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-20 17:19:49.000000000","tz":120},"subject":"Debian api_uwsgi_config: Switch to :ini_setting","message":"Debian api_uwsgi_config: Switch to :ini_setting\n\nOn each puppet run, I\u0027m seeing:\nvalue changed [\u002712\u0027] to \u002712\u0027 (corrective)\n\nTherefore, this patch switches from :openstackconfig to :ini_setting\nto make the 2nd run not changing anything.\n\nChange-Id: Iadf4f00adbb35b832fce7e539e8d76aaecab9acd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/754289fd9d596edc4e25e58ad9edf6399e5ea900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/754289fd9d596edc4e25e58ad9edf6399e5ea900"}]},"branch":"refs/heads/master"},"d3827db1a1e4a1a897a242b7ea481c0e5d69751a":{"kind":"REWORK","_number":2,"created":"2021-04-21 16:13:47.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/72/787272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/72/787272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/72/787272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/72/787272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/72/787272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/72/787272/2"}}},"commit":{"parents":[{"commit":"d0ce60cc3616f4bb8ea0b6fefe1bfb5a3e904320","subject":"Merge \"Add support for panko_api_uwsgi_config in Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/d0ce60cc3616f4bb8ea0b6fefe1bfb5a3e904320"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-20 17:19:49.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-21 16:13:36.000000000","tz":120},"subject":"Debian api_uwsgi_config: Switch to :ini_setting","message":"Debian api_uwsgi_config: Switch to :ini_setting\n\nOn each puppet run, I\u0027m seeing:\nvalue changed [\u002712\u0027] to \u002712\u0027 (corrective)\n\nTherefore, this patch switches from :openstackconfig to :ini_setting\nto make the 2nd run not changing anything.\n\nChange-Id: Iadf4f00adbb35b832fce7e539e8d76aaecab9acd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/d3827db1a1e4a1a897a242b7ea481c0e5d69751a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/d3827db1a1e4a1a897a242b7ea481c0e5d69751a"}]},"branch":"refs/heads/master"},"84bba8a866df95455dcd7ee7df37a692847553f4":{"kind":"REWORK","_number":3,"created":"2021-04-21 16:15:15.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/72/787272/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/72/787272/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/72/787272/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/72/787272/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/72/787272/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/72/787272/3"}}},"commit":{"parents":[{"commit":"d0ce60cc3616f4bb8ea0b6fefe1bfb5a3e904320","subject":"Merge \"Add support for panko_api_uwsgi_config in Debian\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/d0ce60cc3616f4bb8ea0b6fefe1bfb5a3e904320"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-20 17:19:49.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-21 16:15:03.000000000","tz":120},"subject":"Debian api_uwsgi_config: Switch to :ini_setting","message":"Debian api_uwsgi_config: Switch to :ini_setting\n\nOn each puppet run, I\u0027m seeing:\nvalue changed [\u002712\u0027] to \u002712\u0027 (corrective)\n\nTherefore, this patch switches from :openstackconfig to :ini_setting\nto make the 2nd run not changing anything.\n\nChange-Id: Iadf4f00adbb35b832fce7e539e8d76aaecab9acd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/84bba8a866df95455dcd7ee7df37a692847553f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/84bba8a866df95455dcd7ee7df37a692847553f4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
