)]}'
{"id":"openstack%2Fpuppet-panko~791890","triplet_id":"openstack%2Fpuppet-panko~master~I1660e839a5ca63d982ed5dcb180068053bda081f","project":"openstack/puppet-panko","branch":"master","topic":"retire_panko","hashtags":[],"change_id":"I1660e839a5ca63d982ed5dcb180068053bda081f","subject":"Retire puppet-panko.","status":"MERGED","created":"2021-05-18 06:47:49.000000000","updated":"2021-06-09 16:11:51.000000000","submitted":"2021-06-09 16:11:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"791890-retire_panko","meta_rev_id":"67325b49ec31304e357091c51de4bbac008d54f0","_number":791890,"virtual_id_number":791890,"owner":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-09 16:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-09 16:06:51.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-06-09 16:10:35.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-06-09 16:10:35.000000000","_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-18 07:05:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-18 09:21:18.000000000","updated_by":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-06-09 16:10:35.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"ad5ba01ff3b3bbafa27a1e62745d4fb1b97c0948","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-18 06:47:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97200aa315e5602fc76b0361f4a9567a74a1427b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 07:05:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/2a2ddcde9ced4ceb810b58c9c95875c0 : FAILURE in 3m 17s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/338b510a0f9647599168c6990acc2c36 : FAILURE in 3m 16s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/527acc6d46cf4119955af7811dba4da6 : FAILURE in 2m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc47b8d9a77c49a4ad4331e9d304d3dc : FAILURE in 2m 58s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0491d8930120413485def57fb956389b : FAILURE in 6m 33s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b42bae134804fc580ed8b220ef9a7bf : FAILURE in 8m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72524d514c1b4128a704d780ce030912 : FAILURE in 7m 23s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1d4001c14b6440198105e0188f365aa : FAILURE in 6m 06s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64b779ca65054158b3f93a142a330408 : FAILURE in 6m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fe2c0eb10054f8e902c6332f62d9f33 : FAILURE in 7m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/66857d0545c743ccb4b4e8737e4fd38f : FAILURE in 13m 46s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5eb56c5b7525424d9b1137e1093336c8 : FAILURE in 13m 17s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c083f1babc74319b4ec717ee9615df7 : FAILURE in 8m 09s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f561b7f7898479097bb59a15a662383 : FAILURE in 6m 15s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a3c736f9fbf4028b305dd09709ecd0e : FAILURE in 8m 17s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a7ba0f557734b1e91412b9d153e2e09 : FAILURE in 6m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bea2a5e694e040e18e56a0c3ad3bc08a : FAILURE in 5m 27s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fd505383bdf4cf1833035d728134d1c : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"59ccd446d60b37b77126b0d848553b85c13f6fdb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-18 08:14:02.000000000","message":"Topic retire_puppet_panko removed","accounts_in_message":[],"_revision_number":1},{"id":"c3cd886476cac1733b3bbdb855bbb8ad144689cb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-18 08:14:11.000000000","message":"Topic set to retire_panko","accounts_in_message":[],"_revision_number":1},{"id":"ebd26114b34a8399c3760e65a2656da27a49f0dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-18 09:21:05.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6f28c6dac3fdabe2037b842ba0fb8c2ec3c30c85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 09:37:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/2d8a361fe51945d3886f426ae2d4eabe : FAILURE in 2m 41s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6192f23e2ee44c43b14225b8b522adcb : FAILURE in 2m 31s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/399ac848599943c98fba3d1fb6b94397 : FAILURE in 2m 35s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b9b7f5043084354b25b248338e9d60c : FAILURE in 2m 27s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/91d6d6419f4a45079a3dba74688d75c8 : FAILURE in 7m 20s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1f15d4d19dd479b80dabad4c0d14902 : FAILURE in 7m 39s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/310b8bf71c0941f4b9de4fd33017f0a2 : FAILURE in 5m 07s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab89440f0d70499b9cab27a9e6308d9c : FAILURE in 5m 57s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e5e3a3105414b23a7949d67a8ecb086 : FAILURE in 9m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f65ba0d62c4f4395920c5982b7bed36d : FAILURE in 9m 26s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b9d9a953883492ebcea4795d199b2c1 : FAILURE in 11m 10s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc37fb582a064bf0a470970c12d3f7b7 : FAILURE in 11m 37s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bcf16a6dbfc743d0bd26ff4886109bb4 : FAILURE in 8m 04s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f589fff2336645c784cad2d4bce4bdd2 : FAILURE in 5m 44s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/6db0ddba23b04d0b9785bc082d3f652a : FAILURE in 8m 19s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/565fb66d78844a7386e1a541f917a7fc : FAILURE in 5m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/149cc97c10b14a62bb428a4b533de5e6 : FAILURE in 5m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c238f2d9c0844c79b33090b95fe3def8 : FAILURE in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b51d6d5d447220cfe9462b884fad2ecda4f7bba5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-05-18 09:52:39.000000000","message":"Patch Set 2: Code-Review-1\n\nThank you for submitting this :-)\n\nI\u0027ve quickly checked current puppet repos but Panko is still used in integration tests.\nWe should remove all reference to puppet-panko before merging this.\n https://github.com/openstack/puppet-openstack-integration/blob/master/manifests/panko.pp","accounts_in_message":[],"_revision_number":2},{"id":"a2530ec84c0f02bcb69b8dcf7a6e5095de56149e","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-18 10:18:48.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e Thank you for submitting this :-)\n\u003e \n\u003e I\u0027ve quickly checked current puppet repos but Panko is still used in integration tests.\n\u003e We should remove all reference to puppet-panko before merging this.\n\u003e  https://github.com/openstack/puppet-openstack-integration/blob/master/manifests/panko.pp\n\nThank you, good catch. I definitely agree here with you.","accounts_in_message":[],"_revision_number":2},{"id":"47168fe3c1f7e6ec7b9d39fa74a3d257471ada15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-27 09:41:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b8d25f9559c83e81a68c977bf636fe3e16ff9a86","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-27 09:42:35.000000000","message":"Topic retire_puppet_panko removed","accounts_in_message":[],"_revision_number":3},{"id":"4308b0964a04c8b181ba9bc219b5a7f6ea87b0ef","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-05-27 09:42:41.000000000","message":"Topic set to retire_panko","accounts_in_message":[],"_revision_number":3},{"id":"f978da08373c3a055d33d68314ce41bdab9b4d83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 09:59:34.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/323a927a481f4aa99df35899bcf8aa9a : FAILURE in 2m 38s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af6ad5209bcd4bbbb7506771d6836cb2 : FAILURE in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f85cd147c28d4478918ce4d9c84b10a1 : FAILURE in 2m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faa4ed82e0c847fb958cf2c9de4aa66c : FAILURE in 3m 05s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5e5cd3e853e455abf3b6333f6bd3336 : FAILURE in 8m 55s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/057d09a4ba45422c83d28949e698d762 : FAILURE in 9m 40s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f9de85d51cc4029a6fcd23dd44113b9 : FAILURE in 8m 18s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a521c74117a0439ca7679e3c07bb6a92 : FAILURE in 6m 19s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b8c91078dce485fb0bd51fbeed248a3 : FAILURE in 8m 23s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17682ab16ce6470fba24132b6791b2f1 : FAILURE in 7m 56s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b03d2bbf1b084a299a86f58634511218 : FAILURE in 13m 45s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c20f29696dcd484bb210c5e680a44673 : FAILURE in 10m 01s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f055a01120d47968f8f8310fdb58470 : FAILURE in 9m 27s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c410331a694941d2b095e6778977d5fc : FAILURE in 7m 18s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/387b5eed0318445e9430c2d5edab5a60 : FAILURE in 9m 17s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ead2ad065a04a768d3faa48b26359c3 : FAILURE in 5m 40s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed7479b0bd1c4fb88132e25c6ffa6bbe : FAILURE in 7m 26s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9edbd546cd64d548c2aaf06ea1aa44f : FAILURE in 3m 13s","accounts_in_message":[],"_revision_number":3},{"id":"69f8375389dab9f35155c8ec50bc9fa255613bd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-02 06:44:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d54a5d043e4af60f35f10fb30dbff41ec0b78b76","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-02 06:51:43.000000000","message":"Patch Set 4:\n\npanko and pankoclient are gone. \n\nThere is a follow-up patch to re-enable ci jobs in stable branches https://review.opendev.org/c/openstack/project-config/+/793891 which depends on this patch here.","accounts_in_message":[],"_revision_number":4},{"id":"5fa44bab4f63fc3a3cc11f225b5326ed080db533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 07:21:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- noop https://zuul.opendev.org/t/openstack/build/8d3c07c2e0a747d0b63a4db3f11b1221 : SUCCESS in 0s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69598352820e410895979a0b4ede23b3 : FAILURE in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/223f2f36f95e48e69a03f8eb11ea6fed : FAILURE in 3m 21s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b6500a984f24c13a21d98b698234058 : FAILURE in 3m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/872b3b55557c4b1f821ab9c1a7db5e00 : FAILURE in 11m 28s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/9db3b8a63d8a4cf89a876043f2b2fedf : FAILURE in 8m 43s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88a42d5dfb204370bbd4c6016b26cec9 : FAILURE in 8m 16s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50f9af0bed264f74878d75e4b9ca2e64 : FAILURE in 8m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f334f6e0c35f4df685ade681d0911f7e : FAILURE in 8m 51s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/402612fc687c42b58b9ce53acbc4f515 : FAILURE in 13m 12s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7456625f1a4d4ba5a8ba9be1aed8f1ae : FAILURE in 16m 23s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/62c219ee948743b7a4c8039631040d23 : FAILURE in 13m 12s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3a744c149e14b1d9e887cfd80740b8f : FAILURE in 9m 11s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9df997c7b8e419c898295313c21e53f : FAILURE in 6m 46s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/23ad47590c8946fc937847265c0bf55c : FAILURE in 10m 26s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0bcbc65b8354612b5c74adad9d72936 : FAILURE in 7m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c585c4e2e6e49ac87fe1c76791ffc97 : FAILURE in 7m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64d5957ce5424e07adeb01e869817000 : FAILURE in 3m 57s","accounts_in_message":[],"_revision_number":4},{"id":"c5fe6b4a5d3b44dde1699136a90edded6ba7bd30","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-02 07:58:27.000000000","message":"Patch Set 4: Code-Review-1\n\nWe should merge \n https://review.opendev.org/c/openstack/puppet-panko/+/791916\nand\n https://review.opendev.org/c/openstack/puppet-panko/+/794119\nfirst.","accounts_in_message":[],"_revision_number":4},{"id":"adc4c948f2f5a5e395cb0ba884d798fe1af201d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-03 08:27:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f10cd6fb6be4907a275183681b34ad0b452704a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 09:09:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- noop https://zuul.opendev.org/t/openstack/build/be9ae70035e840ecafdc234c8c630879 : SUCCESS in 0s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4bd3a2cc95a491f802ba26c65e3cc80 : FAILURE in 3m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e044f45ce6a44740bc7dc9d710995640 : FAILURE in 2m 53s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41fea28da5e8476f8193978808878f6e : FAILURE in 2m 46s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/83399b7f84c0491bb510a3802dbd4e95 : FAILURE in 9m 25s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/229ce8debb7349aba003730b68a072c3 : FAILURE in 8m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b649ffce3882428384aa8888d1d79035 : FAILURE in 6m 38s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac4d2e28c0044bb59109ed5fd4668de9 : FAILURE in 6m 28s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/892f9e8dfd3344ff99756625da9aa442 : FAILURE in 8m 04s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51d6ec3d6d79483499d2c11f184ac1e9 : FAILURE in 8m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/99aa71f9ff78462aa8427ae8a9c18b38 : FAILURE in 14m 49s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6dbf3705d8c8432c9d5322b537d81403 : FAILURE in 13m 21s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b093313193bf4d128b4743973497f8c9 : FAILURE in 10m 12s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/233834e528ff441da6bf03358224434c : FAILURE in 6m 33s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/cccbd2cad3a54e40af2eb1e0ad000575 : FAILURE in 7m 53s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad6fc0dcce4544ebad2cb79ffabf1a08 : FAILURE in 6m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1da9bdbca9c44a7b2dbfd9f98aba45d : FAILURE in 7m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec7b949d65c5464690cfd33dcc74c471 : FAILURE in 3m 50s","accounts_in_message":[],"_revision_number":5},{"id":"6232e9d400e83c9f977ae8eb7aee4f16dd604681","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-04 10:34:27.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"b5a4292b17f8b7f1a05f3dac88d026e4e66204c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 10:54:26.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- noop https://zuul.opendev.org/t/openstack/build/7e23fc4db6f94d17b20484e78f26681f : SUCCESS in 0s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c906f2f836c14f4281bd4a5e7f20f143 : FAILURE in 2m 51s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e0edd1e74b04337bd554ef9e1e3c1eb : FAILURE in 2m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aac898097b4d49b6998abd59b4399ddd : FAILURE in 2m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/31d7e6a62e8344658e7c88f8a1d097cd : FAILURE in 7m 07s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/66a041de7c6443e5ae494ee1cc966606 : FAILURE in 7m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/643c8dc89d824fceb33c1a023bc848a2 : FAILURE in 5m 24s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9d0fd946801420b894ae7a06cbfbb8d : FAILURE in 4m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9c8a4e3004145e2a00da04722caca8e : FAILURE in 5m 33s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be3179e3e64f43c58c32c1b7676e34d9 : FAILURE in 6m 46s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cab1cc5f01b340758d96fe2d7d6f3034 : FAILURE in 13m 07s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2ba7d7ad1094c41809859aa41f2ddd2 : FAILURE in 11m 24s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2cf1deafb11445399bb36b9c90860e51 : FAILURE in 7m 39s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2daf7ec4ec604015a1a1c755e18c67c0 : FAILURE in 5m 57s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/55fb590f573748ea910057a9e62247a0 : FAILURE in 6m 37s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d82c6b9ccb294885bf9eaa3147d2403e : FAILURE in 4m 32s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ada40fa9eaac4111b71be6b34e61f010 : FAILURE in 4m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2288a85e099446ba2833f20fbca6486 : FAILURE in 3m 06s","accounts_in_message":[],"_revision_number":5},{"id":"31c7872923bb10f8359f5c653e624f973c03cdde","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-05 23:18:18.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f91a49305123df49e61a8fe22b22041b1fa41943","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-07 06:25:10.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3d3a875483003f6b52bd3e205e799359f3810f11","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-07 06:25:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4dce7bedc52f40df22bb454dc2cdc93d38000122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-07 06:51:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- noop https://zuul.opendev.org/t/openstack/build/10da5dd725184d84997ec5af2d588883 : SUCCESS in 0s\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65474a62bf414e5c9234c9fe94e54068 : FAILURE in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89d64c7d974d40debedfa08ee3de52c1 : FAILURE in 3m 41s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e05d1b897eb48b88289f77cd93e026a : FAILURE in 3m 55s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/35d45e68112440dd89886544119527dd : FAILURE in 9m 03s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/a64ac63b3fc6446fb40154dbb6fbfdb2 : FAILURE in 8m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/423ed6a9e58d4abbb0c4cece5d0b8c8c : FAILURE in 6m 46s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/248eb4b70ddb4497ae29c4baafcea5cd : FAILURE in 7m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/877355532c8647f7921e864c765cf4bf : FAILURE in 7m 38s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9103f518c2894fbfb451b2faf68749fd : FAILURE in 8m 32s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8187134a39954dc8b7e2dddc9d792004 : FAILURE in 14m 09s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/35b1b3e101cb42f4b07dc85c48110cdd : FAILURE in 14m 08s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3bd39a0ffe3413cbfdc888d17578e4a : FAILURE in 9m 33s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f6f8fc5a88844598ed13c7d2765f73e : FAILURE in 8m 59s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/b272a2dbf44e4fe8bf50fbebc1df34bd : FAILURE in 9m 00s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f26bf34689f34388b31c1f696b98d1c3 : FAILURE in 5m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2685ad516864de6a6bda0ac5ae1c8bc : FAILURE in 6m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bbdc9268abd48d79347ff28afe74796 : FAILURE in 4m 21s","accounts_in_message":[],"_revision_number":6},{"id":"53bfa732e52ca4161ccab183ff288f1bcebbe2f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-08 14:45:06.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5385b1b24fb4e18a19efdb4d527835c583b7adf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 14:47:43.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/5a1a329bddc1422f8582bf05a3faee76 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":7},{"id":"fb6772db1379f7a161c564d28b6a7df7692ba93a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-08 14:55:23.000000000","message":"Patch Set 7: Code-Review-1\n\nJust one thing. Please remove .gitignore as well. I\u0027ll move the change to p-o-i forward tomorrow if we hear no negative response, then we can move this ahead as well.","accounts_in_message":[],"_revision_number":7},{"id":"0777abbba75a884c6321316f58c27695e086a4ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-09 06:22:55.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"211fcf3baf55f81f1254436be720232ad1f69fe2","author":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"date":"2021-06-09 06:23:22.000000000","message":"Patch Set 8:\n\n\u003e Patch Set 7: Code-Review-1\n\u003e \n\u003e Just one thing. Please remove .gitignore as well. I\u0027ll move the change to p-o-i forward tomorrow if we hear no negative response, then we can move this ahead as well.\n\nDone.","accounts_in_message":[],"_revision_number":8},{"id":"98dd9ccf034e36998639ad18cd91666f37709ba1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 06:23:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/4809cbf0862b4e7dbf937304eb604806 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":8},{"id":"52abd88372a2697d240fe483b5d90313cc303a75","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-09 06:29:50.000000000","message":"Patch Set 8: Code-Review+2\n\nThanks !","accounts_in_message":[],"_revision_number":8},{"id":"b451beb72e1db63035544a5c549c1398f2681295","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-09 06:30:53.000000000","message":"Patch Set 8: Code-Review-1\n\nOops sorry but it seems that some removal patch was blocked by gate failure.\nLet me put -1 until these are rechecked and merged.","accounts_in_message":[],"_revision_number":8},{"id":"bb7cdfa562639305ad22d22b25766b36086547bf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-09 16:06:51.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"9b69243444458a4aceca76e5c1d2cf4ee2d42b05","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-09 16:10:35.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"6b0f2439887d95a6b4294c93b8ae3f6a7290d57a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 16:11:34.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"a18414900bc69add151a8a0017dae46cac2b937a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 16:11:46.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/6985a84c110b416f87ddc976f642cc46 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":8},{"id":"67325b49ec31304e357091c51de4bbac008d54f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 16:11:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"5a97842973a94efb8094ca7dc55377dfcec7571d","revisions":{"2ca1a031918732ad9997eee33d46c5ffcae334c2":{"kind":"REWORK","_number":1,"created":"2021-05-18 06:47:49.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/90/791890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/90/791890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/1"}}},"commit":{"parents":[{"commit":"ef45d33876cfc5489f52cdf9a1e8625b4678d52c","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ef45d33876cfc5489f52cdf9a1e8625b4678d52c"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 06:46:59.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 06:47:47.000000000","tz":120},"subject":"Retire puppet-panko.","message":"Retire puppet-panko.\n\nPanko itself is no longer maintained.\n\nChange-Id: I1660e839a5ca63d982ed5dcb180068053bda081f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/2ca1a031918732ad9997eee33d46c5ffcae334c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/2ca1a031918732ad9997eee33d46c5ffcae334c2"}]},"branch":"refs/heads/master"},"ab8de6ff51719aaa6d3bc50b5dc58acf29beea4f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-05-18 09:21:05.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/90/791890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/90/791890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/2"}}},"commit":{"parents":[{"commit":"ef45d33876cfc5489f52cdf9a1e8625b4678d52c","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ef45d33876cfc5489f52cdf9a1e8625b4678d52c"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 06:46:59.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 09:20:53.000000000","tz":0},"subject":"Retire puppet-panko.","message":"Retire puppet-panko.\n\nPanko itself is no longer maintained.\n\nDepends-on: https://review.opendev.org/c/openstack/project-config/+/791905\n\nChange-Id: I1660e839a5ca63d982ed5dcb180068053bda081f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ab8de6ff51719aaa6d3bc50b5dc58acf29beea4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ab8de6ff51719aaa6d3bc50b5dc58acf29beea4f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e847c700a892ca7af55713b8ecde4834ea227b08":{"kind":"REWORK","_number":3,"created":"2021-05-27 09:41:17.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/90/791890/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/90/791890/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/3"}}},"commit":{"parents":[{"commit":"ef45d33876cfc5489f52cdf9a1e8625b4678d52c","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ef45d33876cfc5489f52cdf9a1e8625b4678d52c"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 06:46:59.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-27 09:40:41.000000000","tz":120},"subject":"Retire puppet-panko.","message":"Retire puppet-panko.\n\nThe contents of this repository are still available in the Git source\ncode management system. To see the contents of this repository before it\nreached its end of life, please check out the previous commit with\n\"git checkout HEAD^1\".\n\nFor any further questions, please email\nopenstack-discuss@lists.openstack.org or join #puppet-openstack on OFTC.\n\nChange-Id: I1660e839a5ca63d982ed5dcb180068053bda081f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/e847c700a892ca7af55713b8ecde4834ea227b08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/e847c700a892ca7af55713b8ecde4834ea227b08"}]},"branch":"refs/heads/master"},"65c47746e05a8aa8b7cca905e5b6d972a6dce639":{"kind":"REWORK","_number":4,"created":"2021-06-02 06:44:26.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/90/791890/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/90/791890/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/4"}}},"commit":{"parents":[{"commit":"d66298deaf3f7aadb2f42d1b27622f9fbb4a1f32","subject":"Prepare Xena M1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/d66298deaf3f7aadb2f42d1b27622f9fbb4a1f32"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 06:46:59.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-06-02 06:44:14.000000000","tz":120},"subject":"Retire puppet-panko.","message":"Retire puppet-panko.\n\nThe contents of this repository are still available in the Git source\ncode management system. To see the contents of this repository before it\nreached its end of life, please check out the previous commit with\n\"git checkout HEAD^1\".\n\nFor any further questions, please email\nopenstack-discuss@lists.openstack.org or join #puppet-openstack on OFTC.\n\nChange-Id: I1660e839a5ca63d982ed5dcb180068053bda081f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/65c47746e05a8aa8b7cca905e5b6d972a6dce639"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/65c47746e05a8aa8b7cca905e5b6d972a6dce639"}]},"branch":"refs/heads/master"},"c687af399a9e8d15a5d49ff0d3941dd96b3be81d":{"kind":"REWORK","_number":5,"created":"2021-06-03 08:27:44.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/90/791890/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/90/791890/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/5"}}},"commit":{"parents":[{"commit":"03688e28410d465a4069d553b7da02bfa228ad7f","subject":"Merge \"Stop acceptance tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/03688e28410d465a4069d553b7da02bfa228ad7f"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 06:46:59.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-06-03 08:27:31.000000000","tz":120},"subject":"Retire puppet-panko.","message":"Retire puppet-panko.\n\nThe contents of this repository are still available in the Git source\ncode management system. To see the contents of this repository before it\nreached its end of life, please check out the previous commit with\n\"git checkout HEAD^1\".\n\nFor any further questions, please email\nopenstack-discuss@lists.openstack.org or join #puppet-openstack on OFTC.\n\nChange-Id: I1660e839a5ca63d982ed5dcb180068053bda081f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/c687af399a9e8d15a5d49ff0d3941dd96b3be81d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/c687af399a9e8d15a5d49ff0d3941dd96b3be81d"}]},"branch":"refs/heads/master"},"b4937055e4cfa564eed375ebe161295311e3f79e":{"kind":"REWORK","_number":6,"created":"2021-06-07 06:25:10.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/90/791890/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/90/791890/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/6"}}},"commit":{"parents":[{"commit":"03688e28410d465a4069d553b7da02bfa228ad7f","subject":"Merge \"Stop acceptance tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/03688e28410d465a4069d553b7da02bfa228ad7f"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 06:46:59.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-06-07 06:24:43.000000000","tz":120},"subject":"Retire puppet-panko.","message":"Retire puppet-panko.\n\nThe contents of this repository are still available in the Git source\ncode management system. To see the contents of this repository before it\nreached its end of life, please check out the previous commit with\n\"git checkout HEAD^1\".\n\nFor any further questions, please email\nopenstack-discuss@lists.openstack.org or join #puppet-openstack on OFTC.\n\nChange-Id: I1660e839a5ca63d982ed5dcb180068053bda081f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/b4937055e4cfa564eed375ebe161295311e3f79e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/b4937055e4cfa564eed375ebe161295311e3f79e"}]},"branch":"refs/heads/master"},"9606a961d914d53e8bea3c989efa582550ddd98b":{"kind":"REWORK","_number":7,"created":"2021-06-08 14:45:06.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/90/791890/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/90/791890/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/7"}}},"commit":{"parents":[{"commit":"03688e28410d465a4069d553b7da02bfa228ad7f","subject":"Merge \"Stop acceptance tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/03688e28410d465a4069d553b7da02bfa228ad7f"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 06:46:59.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-06-08 14:44:54.000000000","tz":120},"subject":"Retire puppet-panko.","message":"Retire puppet-panko.\n\nThe contents of this repository are still available in the Git source\ncode management system. To see the contents of this repository before it\nreached its end of life, please check out the previous commit with\n\"git checkout HEAD^1\".\n\nFor any further questions, please email\nopenstack-discuss@lists.openstack.org or join #puppet-openstack on OFTC.\n\nChange-Id: I1660e839a5ca63d982ed5dcb180068053bda081f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/9606a961d914d53e8bea3c989efa582550ddd98b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/9606a961d914d53e8bea3c989efa582550ddd98b"}]},"branch":"refs/heads/master"},"5a97842973a94efb8094ca7dc55377dfcec7571d":{"kind":"REWORK","_number":8,"created":"2021-06-09 06:22:55.000000000","uploader":{"_account_id":4264,"name":"Matthias Runge","email":"mrunge@redhat.com","username":"mrunge"},"ref":"refs/changes/90/791890/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/90/791890/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/90/791890/8"}}},"commit":{"parents":[{"commit":"03688e28410d465a4069d553b7da02bfa228ad7f","subject":"Merge \"Stop acceptance tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/03688e28410d465a4069d553b7da02bfa228ad7f"}]}],"author":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-05-18 06:46:59.000000000","tz":120},"committer":{"name":"Matthias Runge","email":"mrunge@redhat.com","date":"2021-06-09 06:22:45.000000000","tz":120},"subject":"Retire puppet-panko.","message":"Retire puppet-panko.\n\nThe contents of this repository are still available in the Git source\ncode management system. To see the contents of this repository before it\nreached its end of life, please check out the previous commit with\n\"git checkout HEAD^1\".\n\nFor any further questions, please email\nopenstack-discuss@lists.openstack.org or join #puppet-openstack on OFTC.\n\nChange-Id: I1660e839a5ca63d982ed5dcb180068053bda081f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/5a97842973a94efb8094ca7dc55377dfcec7571d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/5a97842973a94efb8094ca7dc55377dfcec7571d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
