)]}'
{"id":"openstack%2Fpuppet-panko~796184","triplet_id":"openstack%2Fpuppet-panko~stable%2Fussuri~I2c7828b0198f51e8b77b83a0f3a793e747da69c8","project":"openstack/puppet-panko","branch":"stable/ussuri","topic":"bug/1930403-stable/wallaby-stable/victoria-stable/ussuri","hashtags":[],"change_id":"I2c7828b0198f51e8b77b83a0f3a793e747da69c8","subject":"Set fixture paths for unit tests","status":"MERGED","created":"2021-06-13 02:49:46.000000000","updated":"2021-06-28 14:02:57.000000000","submitted":"2021-06-28 14:02:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796184-bug/1930403-stable/wallaby-stable/victoria-stable/ussuri","cherry_pick_of_change":794927,"cherry_pick_of_patch_set":1,"meta_rev_id":"6ba0913b04823a8c63d510902e3831d15e8d442f","_number":796184,"virtual_id_number":796184,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-28 14:02:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-28 12:57:07.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-06-28 12:57:07.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-13 03:22:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44a9eeb7504088b4cf561c55076e807e61eb29ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-13 02:49:46.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"cbceb525339105d2092357afabc2814096e3825f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-13 02:58:59.000000000","message":"Topic cherrypick-bug/1930403-stable/wallaby-stable/victoria-sluuvw5e7s removed","accounts_in_message":[],"_revision_number":1},{"id":"4b7bc46c06552f26ee7b0745876c909d710fd32e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-13 02:59:40.000000000","message":"Topic set to bug/1930403-stable/wallaby-stable/victoria-stable/ussuri","accounts_in_message":[],"_revision_number":1},{"id":"11d96bac95cba3e71a9bdfc7ab38f7cc604a0b6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-13 03:22:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e72264c1e56a47188f4aec493579d6ca : SUCCESS in 3m 51s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac3b4c60a79441ddb2769df3f46f0c53 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e530702854d4155a5038137799bf207 : SUCCESS in 3m 39s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/26092df483a2446dba7895f67e0604c3 : SUCCESS in 13m 19s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/fc03b59d26374323ab6da9ae15bfb37b : SUCCESS in 14m 00s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a253744532641f69357a12221459c03 : SUCCESS in 11m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f20d5a4d02954562ab778ea530e36635 : SUCCESS in 8m 30s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/d19c750138314a4a9b6b610d11053c61 : FAILURE in 17m 01s\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c99b890cd3b49c7be2f7471c61801ba : FAILURE in 24m 55s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60de889b6be24589bc4c043f8bccf571 : FAILURE in 14m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"87acdab4593eddc69a50ed9d6389928a3198dd40","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 12:57:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0865c072374b7e7dd76ce52d1f04a03385a38011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 13:34:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c77c611a1ad4b7193c832b33bac1cb9 : SUCCESS in 5m 37s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/380f6cddad514adaad879700377767a7 : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0e9139d0630419c959a7d1e4c8d75b6 : SUCCESS in 5m 12s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d33938c2300a4507bf53e53e858dc8c7 : SUCCESS in 17m 18s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/2005384b73e24bd29529086faad35e83 : SUCCESS in 16m 48s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e90a1a4f32cc47378786cb19829276d0 : SUCCESS in 10m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b130a69c956c47019d2df3384b6a0c08 : SUCCESS in 10m 11s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/dcbc98a95e464e1b9005cb992997cbf7 : FAILURE in 22m 14s (non-voting)\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1b5841cca5c4920abbd974623919ab7 : FAILURE in 28m 46s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16f066341d2246bcaa138a35ac25aa63 : FAILURE in 16m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a730bcd2e5b5a36dc844d502b967e05cdb24b131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 13:35:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71f79c8dc14f8f4f1329678ed693c718f281619f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 14:02:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/faff00c2bf6b4baba69d8adb3df06df1 : SUCCESS in 6m 25s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/568b37efdaa44dea91a5cf5a111db524 : SUCCESS in 5m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/845b6085a1ae4f15bd138e24ec06878b : SUCCESS in 5m 04s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/049cb1051105408f8434c3723558cce4 : SUCCESS in 15m 36s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/affa656c203a410084f8c46ee05cae64 : SUCCESS in 14m 51s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17d267c91cc14113a083366acb51abfc : SUCCESS in 10m 15s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/cec783739b2543e4938b1893e53dd9a8 : FAILURE in 26m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6ba0913b04823a8c63d510902e3831d15e8d442f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 14:02:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea3430aa51e6896b534a68ee361316d1a0c88e75","revisions":{"ea3430aa51e6896b534a68ee361316d1a0c88e75":{"kind":"REWORK","_number":1,"created":"2021-06-13 02:49:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/84/796184/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/84/796184/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/84/796184/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/84/796184/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/84/796184/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/84/796184/1"}}},"commit":{"parents":[{"commit":"3555d8304a59a14b3075cd35cc9de64970cbe256","subject":"Add support for the interface parameter in authtoken middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/3555d8304a59a14b3075cd35cc9de64970cbe256"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-02 05:19:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-13 02:49:46.000000000","tz":0},"subject":"Set fixture paths for unit tests","message":"Set fixture paths for unit tests\n\nThis change defines manifest_dir and module_path expicitly in unit\ntests so that modules installed under fixtures directory is properly\nloaded.\n\nCloses-Bug: #1930403\nChange-Id: I2c7828b0198f51e8b77b83a0f3a793e747da69c8\n(cherry picked from commit bdbc8382d8f58f380ffaa05a6cd4c522366495fc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ea3430aa51e6896b534a68ee361316d1a0c88e75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/ea3430aa51e6896b534a68ee361316d1a0c88e75"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
