)]}'
{"id":"openstack%2Fpuppet-panko~798856","triplet_id":"openstack%2Fpuppet-panko~stable%2Fvictoria~I27cc0bda33bc55b067a25868ae972b204f06da05","project":"openstack/puppet-panko","branch":"stable/victoria","topic":"at-params-stable/victoria","hashtags":[],"change_id":"I27cc0bda33bc55b067a25868ae972b204f06da05","subject":"Add support for the keystone_authtoken/service_type parameter","status":"ABANDONED","created":"2021-06-30 11:29:33.000000000","updated":"2021-06-30 14:25:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":761075,"cherry_pick_of_patch_set":1,"meta_rev_id":"14f6536785f8dbe9fe12d26e2eba6dfc8a37cddc","_number":798856,"virtual_id_number":798856,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-06-30 13:06:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-30 11:29:33.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-06-30 13:06:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a7e23ec88c1860a8204d9290cd3278c879d3a5ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-30 11:29:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dcf6c945b7459ae4161073c50415b9ce0e8ff1c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 13:06:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c91c9fc8e14b43a791d83c3f97a6afda : SUCCESS in 6m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dab1715b919a453db1e8eee7e44252c1 : SUCCESS in 6m 12s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/f57d8447dbde49e2b8c8cc1794ec904e : SUCCESS in 14m 44s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8c538300f6240bea924d0acabd98128 : SUCCESS in 12m 59s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba978563f0324c2e8fb6f21e81f48a25 : SUCCESS in 10m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b38b85659954425a07489f4e3e39131 : SUCCESS in 1h 04m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7c3b16f3006645c2912ed312f9366874 : SUCCESS in 1h 25m 20s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6df4ee327bd4e5ba849fabd51926ced : SUCCESS in 1h 31m 17s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/39ba8dcb68fa40efa71aac8bcc5496e2 : SUCCESS in 21m 52s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/04761436fb3643df83db68f5a2b3ad01 : SUCCESS in 25m 40s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8da68cecd8354f44b26f8b5c31987784 : FAILURE in 20m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/219ae99542c74e95b6ece441bb0e5edd : FAILURE in 5m 10s","accounts_in_message":[],"_revision_number":1},{"id":"da5131d7dbc515e2fa07e2427c6b01123d982f59","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-30 14:06:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8bcf2b0b87b494ade3fceb5c4bd1f97879b08392","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-06-30 14:07:24.000000000","message":"Abandoned\n\npanko is archived","accounts_in_message":[],"_revision_number":1},{"id":"4ce54524e7516711e8da9b3c2bdd6d5ad0896839","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-30 14:22:28.000000000","message":"Patch Set 1:\n\nIt seems releasenotes/source/\u003crelease\u003e.rst does not exist...","accounts_in_message":[],"_revision_number":1},{"id":"14f6536785f8dbe9fe12d26e2eba6dfc8a37cddc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-30 14:25:15.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e It seems releasenotes/source/\u003crelease\u003e.rst does not exist...\n\nAll, ok so a problem here is that we have archived master so releasenotes/source/victoria.rst no longer exists.\nI\u0027m ok with abandoning the change but I have some concern with that behavior which means that we no longer able to add relesae notes for older releases in case we need something documented with bug fixes...","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"354d28f9ef81430b97f906b2bfc62abd363b54dd","revisions":{"354d28f9ef81430b97f906b2bfc62abd363b54dd":{"kind":"REWORK","_number":1,"created":"2021-06-30 11:29:33.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/56/798856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/56/798856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/56/798856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/56/798856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/56/798856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/56/798856/1"}}},"commit":{"parents":[{"commit":"37d6792c833e8b5391fd6e1f79d319e7bc5245ac","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/37d6792c833e8b5391fd6e1f79d319e7bc5245ac"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-03 08:08:45.000000000","tz":540},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.se","date":"2021-06-30 11:29:33.000000000","tz":0},"subject":"Add support for the keystone_authtoken/service_type parameter","message":"Add support for the keystone_authtoken/service_type parameter\n\nChange-Id: I27cc0bda33bc55b067a25868ae972b204f06da05\n(cherry picked from commit e290929657419faed698a6d15b160ea5bd53f4cc)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/354d28f9ef81430b97f906b2bfc62abd363b54dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/354d28f9ef81430b97f906b2bfc62abd363b54dd"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
