)]}'
{"id":"openstack%2Fpuppet-panko~820623","triplet_id":"openstack%2Fpuppet-panko~stable%2Ftrain~I8ae2e126048c130f7c498d5ec0e790bc0dd42c1e","project":"openstack/puppet-panko","branch":"stable/train","topic":"at-interface-stable/ussuri-stable/train","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-07-20 12:56:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8ae2e126048c130f7c498d5ec0e790bc0dd42c1e","subject":"Add support for the interface parameter in authtoken middleware","status":"ABANDONED","created":"2021-12-07 14:27:33.000000000","updated":"2022-07-20 12:56:36.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":765275,"cherry_pick_of_patch_set":2,"meta_rev_id":"d1d22fead6614d2a63e30f220852edb8728416de","_number":820623,"virtual_id_number":820623,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-12-07 17:53:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-07 17:53:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3c40fd503c82a322a77da1ec211419d944bac49d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-07 14:27:33.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"adae35820feb034c9607de95377bd9d9f2b853dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 17:53:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc4ec379b9844c7981a88f1c50afbbab : SUCCESS in 5m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/989e4aa1e9bf489c8222c52073a29b5b : SUCCESS in 5m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/87b176bc755743869c0721268efcfaf6 : SUCCESS in 5m 00s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3863495f896441d1b07bf35af734b235 : SUCCESS in 19m 27s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/9416db9eaf574085b882c81ff376dadf : SUCCESS in 11m 58s\n- puppet-openstack-unit-6.4-centos-8-stream https://zuul.opendev.org/t/openstack/build/df2ec9833f9846e4967e6210e7cead64 : SUCCESS in 12m 03s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17b277b1047b4c9da682ce2daaefa54a : SUCCESS in 10m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc8d57504a694d84b680b92a8ebf6cd1 : SUCCESS in 14m 11s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/14ec4114b3734658b709fd234056df83 : SUCCESS in 49m 46s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/a527163024454186830db1a94d80a6c5 : SUCCESS in 54m 25s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c64bd630a79245f88992990c990499f2 : SUCCESS in 55m 43s\n- puppet-openstack-integration-5-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5140c6efa79649068ff5ab4249ba1c8a : SUCCESS in 49m 39s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f7ee9bcf22ae4f48975d86e5c4c2ebcb : SUCCESS in 19m 19s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/458f0eb84bfa4919b1a9726bbe9f7d87 : FAILURE in 9m 25s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1236b274a4d3493682c57aa61713258a : FAILURE in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"d1d22fead6614d2a63e30f220852edb8728416de","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-07-20 12:56:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81276b62a944e9d529302513d3a000851c0c71da","revisions":{"81276b62a944e9d529302513d3a000851c0c71da":{"kind":"REWORK","_number":1,"created":"2021-12-07 14:27:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/23/820623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-panko","ref":"refs/changes/23/820623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/23/820623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/23/820623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-panko refs/changes/23/820623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-panko refs/changes/23/820623/1"}}},"commit":{"parents":[{"commit":"da522675438c3edf31998558938ecca0c7895bb5","subject":"Prepare the final stable/train release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/da522675438c3edf31998558938ecca0c7895bb5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-08 23:37:24.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-07 14:27:33.000000000","tz":0},"subject":"Add support for the interface parameter in authtoken middleware","message":"Add support for the interface parameter in authtoken middleware\n\nThis patch adds support for [keystone_authtoken] interface parameter,\nso that operators can define which endpoint should be used by authtoken\nmiddleware.\n\nChange-Id: I8ae2e126048c130f7c498d5ec0e790bc0dd42c1e\n(cherry picked from commit ba317b7ef3c6212cc60473457dae94ca828959bf)\n(cherry picked from commit 3555d8304a59a14b3075cd35cc9de64970cbe256)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/81276b62a944e9d529302513d3a000851c0c71da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-panko/commit/81276b62a944e9d529302513d3a000851c0c71da"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
