)]}'
{"id":"openstack%2Fpuppet-placement~687042","triplet_id":"openstack%2Fpuppet-placement~stable%2Ftrain~Ic164179e383ecefb5201805c1b26e1ffa66c4a80","project":"openstack/puppet-placement","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"Ic164179e383ecefb5201805c1b26e1ffa66c4a80","subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","status":"MERGED","created":"2019-10-07 13:34:51.000000000","updated":"2019-10-10 21:50:56.000000000","submitted":"2019-10-10 21:49:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687042-1570744188400-33a7a8e1","meta_rev_id":"adde885f970eb7903efb5f3fde3ca3057eec958c","_number":687042,"virtual_id_number":687042,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-10-10 19:19:20.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-10 21:49:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-10-07 21:11:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-10-07 22:31:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-07 21:11:19.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-10-07 22:31:10.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2019-10-10 19:19:20.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-10-10 21:49:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c73db28b89e5acd732d3801360858095c490816","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-10-07 13:34:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8162c7eeddaf1fbc307962f2caa445e829b3f75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 13:44:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79f1c0a6b92747d49c44ac534535767c : SUCCESS in 2m 58s","accounts_in_message":[],"_revision_number":1},{"id":"b9d2c2b1085fe414cd5991060512c026c6a2a1ac","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-10-07 21:11:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1faad5f8a0d039d539b1dc66e04bf970fef88ac6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2019-10-07 22:31:10.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"46eaec23d68752ac7fff33262f80444365a12b6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 14:35:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17ec5ad602dcff419681fd3cf24b8de723b78f6e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 15:59:55.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12c8245f9a874f718165327ed268f25e : SUCCESS in 5m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3d7d573862fb43ee997054f4d86fc9e3 : SUCCESS in 6m 01s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b29dec41ca6a4085956a485da42c31ce : SUCCESS in 14m 41s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/85728c265d0b4c2dbf7ee41f2ed7b945 : FAILURE in 56m 17s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8bfd1a8a70bc45ceb6b5cf2eb76b88b3 : SUCCESS in 1h 08m 28s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/136a70aa41bf44f4980446281b37cc2a : SUCCESS in 1h 12m 13s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/315e1526fb894e4b9c7acfab0d9f8ad4 : SUCCESS in 47m 53s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f524b3cdec4849d886ed8ac79d6df7dd : SUCCESS in 10m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cabc127ebdc44ea7bd4eee388aaaab7f : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"751273d8eacae0df4da999eba64ac8dfa746b3e0","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2019-10-10 19:19:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9b73f7b0b561f67214054e70a8c299a37a8abbcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 20:28:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc0e3eb7517442de8173e356e0db9237 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/16c094bdaed545c595c9be3a48ca4a9b : SUCCESS in 4m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c4f29cb1689e4f4aae0a4c892903099b : SUCCESS in 16m 52s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/2a0166c1e6f444bebee59e2a4f4c91ba : SUCCESS in 51m 01s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/504263d91c76465b9ddf86f49784f1e0 : SUCCESS in 56m 37s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/be72d0d34247416bad57cf5da0bf243c : SUCCESS in 1h 00m 09s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/24b521f22e8f43f196fd13b628371004 : SUCCESS in 50m 36s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/7c74aba7b35340b7bf47abff6f1e63a1 : FAILURE in 29m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/78a54fc267794a1382e1921498a6ace5 : FAILURE in 34m 52s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06870a6d10994852b6694c4e136089bc : FAILURE in 20m 08s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/e2496a8acfae42859eb2aac2b4a845fb : FAILURE in 23m 17s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5abccba6071c4881a6c2aee01ba46a8a : SUCCESS in 8m 44s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d72a0425b58741db870ddb2c8aa8f514 : SUCCESS in 5m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b509276b1c984eebaa221b59b41745e6 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":1},{"id":"2df3ebf6299dce29578c3c4ddbc3ad3a77a4fe0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 20:28:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e1a0b22eae375903f0108e41feda959fb9025a52","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 21:49:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fa5c5d8c8c46f8c567faf73028963324822fef16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 21:49:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/131417cfe71b4910b6b64118516abd9f : SUCCESS in 4m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc17203ba8fd4cf7a998c90553bf68c7 : SUCCESS in 4m 00s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4ea6eec8711f4391afd9abf034f73d4e : SUCCESS in 9m 53s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/51cde0ffcf5c4424890ee8031a6a295f : SUCCESS in 51m 59s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/cb93d91c777b4f2f889a9035297c5809 : SUCCESS in 59m 07s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/328724911aae4e33be0e9e2006445f19 : SUCCESS in 1h 09m 11s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/8ff7b943ef2a4af3ba33596a242d234e : SUCCESS in 52m 20s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4ecd3579435d47c28c56025acacaf2b9 : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f86cfad6a8534cdeae1dfef978356175 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"cfdd51042c731a27f5779876cc9c109b54f1a401","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-10 21:50:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f54c7ba42d1430f86de87f15ccfdc7a : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"583347051aba67d5ecde2ce167f2d8c359f855d3","revisions":{"583347051aba67d5ecde2ce167f2d8c359f855d3":{"kind":"REWORK","_number":1,"created":"2019-10-07 13:34:51.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/42/687042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/42/687042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/42/687042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/42/687042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/42/687042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/42/687042/1"}}},"commit":{"parents":[{"commit":"ec1f3a69d06c97cae0b4472f83f32d2f6261a018","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/ec1f3a69d06c97cae0b4472f83f32d2f6261a018"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-07 13:34:49.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-10-07 13:34:49.000000000","tz":0},"subject":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train","message":"Update TOX/UPPER_CONSTRAINTS_FILE for stable/train\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/train branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Ic164179e383ecefb5201805c1b26e1ffa66c4a80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/583347051aba67d5ecde2ce167f2d8c359f855d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/583347051aba67d5ecde2ce167f2d8c359f855d3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
