)]}'
{"id":"openstack%2Fpuppet-placement~745272","triplet_id":"openstack%2Fpuppet-placement~master~I3f669686896b543f751fe2145bdc2f40aaa310bf","project":"openstack/puppet-placement","branch":"master","topic":"placement-port","hashtags":[],"change_id":"I3f669686896b543f751fe2145bdc2f40aaa310bf","subject":"Run placmeent on an independent port by default","status":"ABANDONED","created":"2020-08-07 00:59:08.000000000","updated":"2021-07-12 02:49:03.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"86a4dd8211f33c1fa745a67c640574c588f3d9c6","_number":745272,"virtual_id_number":745272,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-08-31 15:02:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":-2,"date":"2020-08-31 14:23:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-10 12:50:58.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-31 13:02:51.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-08-31 14:19:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-31 15:02:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"91325019dd1b38f7e529f2aaa926a0a1f7730a25","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-07 00:59:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f1155db365c356216760244140b5689cc2399567","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-07 02:58:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82a0788bf1fd4c0390f8e75324a175fd : SUCCESS in 3m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8ac878833f64ebd8c79c611bac2f99e : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b26e597d3d54d5ba61bda2ba4d2c64a : SUCCESS in 4m 01s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/93dd38e9db194653a4f21a1ecf821021 : SUCCESS in 8m 44s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/aa22a689f8604f4b81b4e7534272c7a4 : SUCCESS in 9m 16s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0387b809ec7940ab82bda30d199da266 : SUCCESS in 7m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/661e3c0615df4be9bb17659527e6dfe4 : FAILURE in 6m 06s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1076ae2b3f004f8db44138b9310eeeb3 : SUCCESS in 34m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/776774180efd4c4caabe7286563aee21 : FAILURE in 36m 18s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9daca3abc5846ffa5348f7f0da5ad9f : SUCCESS in 34m 09s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/710421b3bbfa4a80a23250abcf61d569 : SUCCESS in 29m 08s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a9b8935b3dd1446b828bb0d4e0ae3d3e : SUCCESS in 45m 43s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c288d3dd673a473eadd9b8bea16ea4a9 : SUCCESS in 53m 06s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3133a5b7f9054be2b4cc3d0ded0118b8 : SUCCESS in 1h 54m 29s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4c949ba92bcc428990b2f1893bbb6e76 : SUCCESS in 53m 52s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9ce8ae2fc9ac4f25a01d5ed197f8500a : SUCCESS in 8m 27s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/d92676c099764750b483ab03bdad67df : SUCCESS in 7m 50s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/31aca5a1457e4e3ab0a67a5d6cee4eab : SUCCESS in 7m 14s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81ee5b4a217945ff9ca69253258a6270 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6abdb7f238c0c288f51e46f6e73b1724e13e390b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-10 12:50:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2537ea04a2a0b44a77eca8c6ffdb5c4603fed3c7","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-10 13:33:28.000000000","message":"Patch Set 1: Code-Review+2\n\nI think this is valid since installation guides uses port 8778 even though historic discussion and devstack is deploying in port 80 with /placement as URL prefix.\n\nNot doing +w so that others can leave their feedback.","accounts_in_message":[],"_revision_number":1},{"id":"5a295c30fd2745e9f159b7dd0707592ff7ac47fc","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-11 14:46:55.000000000","message":"Patch Set 1: Code-Review+2\n\nmaybe send a note to the ML? though i think most folks are overriding it anyway.  If we get no response next week, let\u0027s merge","accounts_in_message":[],"_revision_number":1},{"id":"0f205899be69c8cb85c3e6d4076b25dde9635c37","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 09:02:40.000000000","message":"Patch Set 1:\n\nWe should just merge this IMO.","accounts_in_message":[],"_revision_number":1},{"id":"08732a32aca34e51f5e1e555f8afa72a5558be23","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-31 13:02:51.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"59313be487b2ce6250d4b92eafb0b5ce6a9da640","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:03:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5429ddd4be32f660dae4fb6c526daea84b996d17","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 13:09:09.000000000","message":"Patch Set 1: Code-Review-2\n\nSorry but let me put -2 atm because I noticed that I should update default endpoint urls, too, otherwise the default doesn\u0027t work...","accounts_in_message":[],"_revision_number":1},{"id":"ed729313187cf04b1b3e8b8cbc7ef642545a2fcb","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-08-31 13:12:54.000000000","message":"Removed Workflow+1 by Alex Schultz \u003caschultz@redhat.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"c90cc622a2c40f452ed4e290592531453124a3e1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 13:46:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"efaafd16fe46c377cccbbaaa1e9f19b10496588a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 13:49:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"218e41a17ba56c83e402ec1ee992977e5fdfa397","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 14:12:40.000000000","message":"Patch Set 3: -Code-Review Workflow-1\n\nDuring updating default endpoint definitions I noticed that placement api has 2 configurations different from the other services.\n\n 1. api_port uses 80 instead of service specific port\n 2. path uses /placement instead of /\n\nIf we follow the installation doc then we need to change both of the above 2 points, but I\u0027m not confident with \"popular\" set up especially about 2. For example TripleO still depend on the default /placement path while it use 8778 to override default port 80.\n\nNow I feel that changing 1 only is incomplete, while changing both seems risky, and it would be better to leave the current defaults and have users to set required parameters likr path if they want to use independent port, instead of changing default to uncertain values which can affect existing deployment.\n\nThe 2 concern with leaving current defaults are\n - the default keystone endpoint doesn\u0027t work on Debian which don\u0027t use httpd to run placement api\n - we still need to maintain special defaults\nbut I don\u0027t think these matters much more compared to possible bad effects caused by this change.\n\nI\u0027ll wait for any feedback with keeping this patch open for review, but I\u0027d tend to say that I should abandon this.(or does it still make sense to change port only ?)","accounts_in_message":[],"_revision_number":3},{"id":"628f6fc80449aeda526c3b82fd0cac0b35264abc","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 14:19:08.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nI\u0027m kind of biased since we\u0027re already using port 8778 and doing use any URL prefix, not sure what the installation guide says about URL prefix? We pretty have it like this since it\u0027s easier to distinct and was configured this way before Placement was standalone from Nova.","accounts_in_message":[],"_revision_number":3},{"id":"8c751e4a75f42ab57767d29c5fd94297169eefcc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 14:23:45.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"327ae0d6c95dfa5601151647cadf7e649226bd9d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 14:46:38.000000000","message":"Patch Set 4:\n\nThe current installation guides suggest tcp/8778 without any prefix. I confirmed that this comes from the default wsgi conf file in rdo[1], I\u0027m still looking for the one in Ubuntu but I expect the same regarding the same description in the doc.\n\n[1] https://github.com/rdo-packages/placement-distgit/blob/rpm-master/placement-api.conf#L8","accounts_in_message":[],"_revision_number":4},{"id":"54cf7aaa8139cf86538dbb2692754b59dbb17e64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 15:02:43.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9ca3c80dd5441ac87a1af0f2a73d1ab : SUCCESS in 4m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17bf8cefe8ca4de8b49aeffb6b281105 : SUCCESS in 5m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b34c0a2b70294b11b0b12cd6c3b59ef6 : SUCCESS in 5m 06s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/442d414c733645268f7abf3ca29ced84 : FAILURE in 9m 53s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/87b992be9c5546179e96125a25f3684f : FAILURE in 11m 04s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d180c3a3c7154aa5904ec49199d43453 : FAILURE in 9m 49s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5db54921e46944e2baed205333e27cb6 : FAILURE in 11m 27s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c17a5e97bfa54d35914a8ea05e922619 : FAILURE in 7m 56s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e6a833648b04e7eb0fb725bb1485c06 : FAILURE in 7m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d6b7c380efb47868bd2470a467fbaca : FAILURE in 6m 14s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/81e0f372d5a74790a9daaeb983e14e76 : FAILURE in 10m 08s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0d7ab2be28844cd2995e6a9ec8e34f6c : FAILURE in 12m 02s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b8257b0e566b4ce59c02d115cde6da16 : FAILURE in 12m 41s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cfe2ac7a83ae429392a553e7b91e1909 : FAILURE in 12m 41s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8e678f2be49a477c87578e325ebc6cb7 : FAILURE in 11m 23s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/101094ca4595455bae838870539cc7f4 : FAILURE in 9m 58s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c757f912f134b50b4582968e72916d2 : FAILURE in 11m 22s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c573095485247a8b120c2107cefa0bf : SUCCESS in 5m 01s","accounts_in_message":[],"_revision_number":4},{"id":"86a4dd8211f33c1fa745a67c640574c588f3d9c6","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-12 02:49:03.000000000","message":"Abandoned\n\nI have restored this work in https://review.opendev.org/c/openstack/puppet-placement/+/800363","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2a02ecbc092480e690738145ad3c856358fe0f97","revisions":{"1b5ed0aa1913837be26735dec7f0317ce73f574f":{"kind":"REWORK","_number":1,"created":"2020-08-07 00:59:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/745272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/72/745272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/1"}}},"commit":{"parents":[{"commit":"57daa72bea427214dfbd3d6fbfa97ed3ab396cef","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/57daa72bea427214dfbd3d6fbfa97ed3ab396cef"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-07 00:57:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-07 00:57:17.000000000","tz":540},"subject":"Change default api_port","message":"Change default api_port\n\nThis patch replaces the current default api port(80) by 8778, because\ntcp/8778 is more commonly used as placement service port.\n\nChange-Id: I3f669686896b543f751fe2145bdc2f40aaa310bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/1b5ed0aa1913837be26735dec7f0317ce73f574f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/1b5ed0aa1913837be26735dec7f0317ce73f574f"}]},"branch":"refs/heads/master"},"e8fb4756f29f6b47eb8eba00a9fd71955d71bbe7":{"kind":"REWORK","_number":2,"created":"2020-08-31 13:46:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/745272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/72/745272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/2"}}},"commit":{"parents":[{"commit":"57daa72bea427214dfbd3d6fbfa97ed3ab396cef","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/57daa72bea427214dfbd3d6fbfa97ed3ab396cef"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-07 00:57:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 13:46:04.000000000","tz":540},"subject":"Change default api_port","message":"Change default api_port\n\nThis patch makes placmenet service use dedicated port, tcp/8778,\ninstead of the default apache port, tcp/80, to follows the way\ndescribed in installation guide[1], and be consistent among all\nservices deployed as wsgi services.\n\n[1] https://docs.openstack.org/placement/latest/install/\n\nDepends-on: https://review.opendev.org/#/c/749058/\nChange-Id: I3f669686896b543f751fe2145bdc2f40aaa310bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/e8fb4756f29f6b47eb8eba00a9fd71955d71bbe7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/e8fb4756f29f6b47eb8eba00a9fd71955d71bbe7"}]},"branch":"refs/heads/master"},"2997d3ca8aca859b16c5a95f5c7239f37581eb46":{"kind":"REWORK","_number":3,"created":"2020-08-31 13:49:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/745272/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/72/745272/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/3"}}},"commit":{"parents":[{"commit":"57daa72bea427214dfbd3d6fbfa97ed3ab396cef","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/57daa72bea427214dfbd3d6fbfa97ed3ab396cef"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-07 00:57:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 13:48:52.000000000","tz":540},"subject":"Run placmeent on an independent port by default","message":"Run placmeent on an independent port by default\n\nThis patch makes placmenet service use dedicated port, tcp/8778,\ninstead of the default apache port, tcp/80, to follows the way\ndescribed in installation guide[1], and be consistent among all\nservices deployed as wsgi services.\n\n[1] https://docs.openstack.org/placement/latest/install/\n\nDepends-on: https://review.opendev.org/#/c/749058/\nChange-Id: I3f669686896b543f751fe2145bdc2f40aaa310bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/2997d3ca8aca859b16c5a95f5c7239f37581eb46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/2997d3ca8aca859b16c5a95f5c7239f37581eb46"}]},"branch":"refs/heads/master"},"2a02ecbc092480e690738145ad3c856358fe0f97":{"kind":"REWORK","_number":4,"created":"2020-08-31 14:23:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/745272/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/72/745272/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/72/745272/4"}}},"commit":{"parents":[{"commit":"57daa72bea427214dfbd3d6fbfa97ed3ab396cef","subject":"Prepare Victoria M2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/57daa72bea427214dfbd3d6fbfa97ed3ab396cef"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-07 00:57:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 14:23:38.000000000","tz":540},"subject":"Run placmeent on an independent port by default","message":"Run placmeent on an independent port by default\n\nThis patch makes placmenet service use dedicated port, tcp/8778,\ninstead of the default apache port, tcp/80, to follows the way\ndescribed in installation guide[1], and be consistent among all\nservices deployed as wsgi services.\n\n[1] https://docs.openstack.org/placement/latest/install/\n\nDepends-on: https://review.opendev.org/#/c/749058/\nChange-Id: I3f669686896b543f751fe2145bdc2f40aaa310bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/2a02ecbc092480e690738145ad3c856358fe0f97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/2a02ecbc092480e690738145ad3c856358fe0f97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
