)]}'
{"id":"openstack%2Fpuppet-placement~748995","triplet_id":"openstack%2Fpuppet-placement~master~If1f63aaf571af497105871da297e1c07d7978c54","project":"openstack/puppet-placement","branch":"master","topic":"validate-db_conn","hashtags":[],"change_id":"If1f63aaf571af497105871da297e1c07d7978c54","subject":"Do not validate database_connection format","status":"MERGED","created":"2020-08-31 08:55:03.000000000","updated":"2020-08-31 15:44:03.000000000","submitted":"2020-08-31 15:44:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748995-1598888643172-210ac784","meta_rev_id":"1d9809c9742cffa52e423c4ea04e01883f90fd1c","_number":748995,"virtual_id_number":748995,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-31 15:44:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-08-31 13:27:20.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-08-31 11:02:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":2,"date":"2020-08-31 09:04:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-08-31 13:27:20.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-31 08:55:35.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-08-31 11:02:28.000000000","updated_by":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"reviewer":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"state":"REVIEWER"},{"updated":"2020-08-31 13:27:20.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-08-31 15:44:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dae94fd57270186ec335d3e47f5f4cdf1294aaff","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 08:55:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca4258e309965d54d1aff812ac9bff2b2388ae9d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-08-31 08:55:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3cd3f385c03b26b72f3f005e3901f85618a767db","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-08-31 09:04:00.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"e53784050a50c65bb836e567a3b7b6ffb6cc9c8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 10:36:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f173f32106fd470aa73dda7ef0e3e119 : SUCCESS in 3m 15s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd81863cd20342ce9c2baf88259ec600 : SUCCESS in 3m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b7d5d4771ba4c83a2dc0e1288247c15 : SUCCESS in 3m 12s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/90324cc2da1a4ba485318267287512ca : SUCCESS in 11m 21s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/48552f9c4acc4fd2a45ed33d0c615521 : SUCCESS in 9m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c9b234614504c3daaf9911fc94f3f14 : SUCCESS in 10m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc9236b219dd49eab55a89a72481bf77 : SUCCESS in 7m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/f7dd8ba4dbf745b5b0fce38b2e58a3e5 : SUCCESS in 57m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15a20e612a87462ca2a6051b411910a9 : FAILURE in 50m 49s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bd2fa46f844493eb881bb12b372621a : SUCCESS in 53m 47s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/62627707aa1d453c9a41c6affd07a77b : SUCCESS in 39m 43s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6fe0c5d26c8c40c7978eca816556a42a : SUCCESS in 1h 21m 57s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/327e196914ef47568353313d4bbc8637 : SUCCESS in 1h 26m 35s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e122c8faef644497a22a34993f245efe : SUCCESS in 1h 28m 27s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d8a0afe8485c40678814c278e435014b : SUCCESS in 55m 28s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/779f7aaa9f344568b10921e52b7f13e1 : SUCCESS in 25m 29s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a939d1a77a904fe58b21ab735857f727 : SUCCESS in 12m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c8a18bc3e5eaa31927bf2c96959a28483b940a1","author":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"date":"2020-08-31 11:02:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"257481d84929d275c596d9b346866a966b9ddeef","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-08-31 13:27:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9b9b313ba9b6e14fda051eefb0404e72998819e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 13:27:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"113ece5648eb654207cdc6b46dd65e6fca994592","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 15:44:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb690b57d68c4f22b2978d65dcecbdde : SUCCESS in 4m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29bf8e2ea992425c8ad6ea3abf27f764 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2e76534abe64ddfafe39049b5993dd3 : SUCCESS in 4m 21s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ab398c0ce1cb4e21b7c96e7fc76f1c2d : SUCCESS in 8m 44s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1f78d75f044e4042b27b08beef7c4949 : SUCCESS in 8m 29s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/adfc0e1e4c58457291ad156f93cb109e : SUCCESS in 7m 04s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3ce54d91abce44759849af022b133bea : SUCCESS in 1h 09m 29s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5e69aa459e014653a5991a3b33b68418 : SUCCESS in 1h 49m 25s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a8b77cf367be4c6bb56510d20c20f177 : SUCCESS in 1h 48m 32s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/517949c4a66e4844871dbb24c481682a : SUCCESS in 55m 33s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/a23f006aff424baaa46527a51074fbd1 : SUCCESS in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c87ff54f99bfecc2bbb350a6425459d8b2c36af1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-31 15:44:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81a77ce901ec519aa81050cd1605dc36419fa066","revisions":{"bb6d91840714687165107965365a4a060412ddcf":{"kind":"REWORK","_number":1,"created":"2020-08-31 08:55:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/748995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/95/748995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/95/748995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/95/748995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/95/748995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/95/748995/1"}}},"commit":{"parents":[{"commit":"34cb28a6f8f5ee41e091003d314ebf281b228b53","subject":"Add mysql_enable_ndb option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/34cb28a6f8f5ee41e091003d314ebf281b228b53"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 08:54:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 08:54:54.000000000","tz":540},"subject":"Do not validate database_connection format","message":"Do not validate database_connection format\n\nCurrently we validate database_connection in 2 layers, each puppet\nmodules and puppet-oslo, however this makes it difficult to maintain\nvalidation pattern because we always need to fix both.\nThis patch removes the validation from each puppet modules so that\nwe need to maitain only one place, puppet-oslo to update validation\nlogic.\n\nChange-Id: If1f63aaf571af497105871da297e1c07d7978c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/bb6d91840714687165107965365a4a060412ddcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/bb6d91840714687165107965365a4a060412ddcf"}]},"branch":"refs/heads/master"},"81a77ce901ec519aa81050cd1605dc36419fa066":{"kind":"NO_CHANGE","_number":2,"created":"2020-08-31 09:04:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/748995/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/95/748995/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/95/748995/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/95/748995/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/95/748995/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/95/748995/2"}}},"commit":{"parents":[{"commit":"34cb28a6f8f5ee41e091003d314ebf281b228b53","subject":"Add mysql_enable_ndb option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/34cb28a6f8f5ee41e091003d314ebf281b228b53"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 08:54:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-08-31 09:00:11.000000000","tz":540},"subject":"Do not validate database_connection format","message":"Do not validate database_connection format\n\nCurrently we validate database_connection in 2 layers, each puppet\nmodules and puppet-oslo, however this makes it difficult to maintain\nvalidation pattern because we always need to fix both.\nThis patch removes the validation from each puppet modules so that\nwe need to maitain only one place, puppet-oslo to update validation\nlogic.\n\nChange-Id: If1f63aaf571af497105871da297e1c07d7978c54\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/81a77ce901ec519aa81050cd1605dc36419fa066"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/81a77ce901ec519aa81050cd1605dc36419fa066"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
