)]}'
{"id":"openstack%2Fpuppet-placement~761076","triplet_id":"openstack%2Fpuppet-placement~master~Ic9df546f304fb0588bd5d18f7979c8df8783e0cb","project":"openstack/puppet-placement","branch":"master","topic":"at-params","hashtags":[],"change_id":"Ic9df546f304fb0588bd5d18f7979c8df8783e0cb","subject":"Add support for the keystone_authtoken/service_type parameter","status":"MERGED","created":"2020-11-03 08:14:15.000000000","updated":"2020-11-07 00:13:49.000000000","submitted":"2020-11-07 00:12:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"761076-1604707964645-96dc3111","meta_rev_id":"36ba5e8cfb9b51b02882cafd5f0a067f67c8cbd3","_number":761076,"virtual_id_number":761076,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-11-04 12:08:19.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-07 00:12:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-11-05 11:13:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-11-06 22:20:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-11-06 22:20:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-05 11:13:29.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-11-06 22:20:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-11-07 00:12:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0a43288a4cc79108dd2a9fabfb0d5ca8158fabd8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-03 08:14:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73ed8fa208966ab306ac3d14ce96356c95f91653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-03 10:06:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b00bdadd075409789558c8425b8da33 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f39f756d8154420b458b260ed642339 : SUCCESS in 3m 24s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d0a8e4ac357241a3b66721ae059a0a62 : SUCCESS in 10m 02s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eff1844c0fe74df08f12dc25663129b3 : SUCCESS in 7m 47s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4dfc6f8cc3dd4d8d93ef9991278054df : SUCCESS in 7m 11s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/c4f781d25e104032bd90fb572ec78905 : FAILURE in 6m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/601298175dde4cf5adb9dc57c42c297c : FAILURE in 5m 55s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6feb1715c9d04649a2b74da6b16d143c : FAILURE in 6m 23s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/0b0383a8eac24f37a3782d95d34c8639 : FAILURE in 6m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/36103a4568a940428d01da72d38fcaaa : SUCCESS in 1h 25m 20s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cfe2d2a9a30c43a8a60fd3dc524f4af2 : SUCCESS in 1h 41m 27s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/80e44df1cc244f84b0c6c37b7355f133 : SUCCESS in 1h 33m 04s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bc6112e7ab6f44e9aed4e3e46d1ba59f : FAILURE in 44m 06s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/9ebee19b3346488c93568950d5930b5d : SUCCESS in 14m 44s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4280f2dda1c54855be6a82537f1c389d : SUCCESS in 16m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2b4a5b3ce504f8a876eea260e5b0162 : SUCCESS in 3m 16s","accounts_in_message":[],"_revision_number":1},{"id":"14260f197fa20e03d4d64378ccfab8b8f00a28b8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-11-04 12:08:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8038bcfc2bc357ecf39e51acca86a780763ff593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 13:54:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8424f77e21bb4bb9afab126ea3069412 : SUCCESS in 3m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8b9fa69b11a44f758ddb523597699162 : SUCCESS in 2m 42s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cf04b22fa8d14d1d90efaa83c6119309 : SUCCESS in 15m 25s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9be5df9bf4d34ea488ae8d4773ec3ec3 : SUCCESS in 6m 44s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/065c6a5eca7f443984bfd84f624f600d : SUCCESS in 5m 38s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/1514d9a252f84c4ba3ad242b34a883b5 : FAILURE in 6m 03s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d398b39482e4e8e8a21cc66b6261bcc : FAILURE in 7m 49s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb3d5cecd0684086b0b3d898b731c24a : FAILURE in 6m 26s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/5dfcb4629ae74bccb141493cdf7b3639 : FAILURE in 6m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/41010378979c43609e2fdbadd2f770d3 : SUCCESS in 1h 22m 56s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7e092a7d94234327abb1c3020bd0fb4f : SUCCESS in 1h 35m 47s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6b2d7ada77da442ba4fb14f8d27ebae3 : SUCCESS in 1h 23m 55s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8a1f63cb4dcf45829f8cb738220ef181 : SUCCESS in 56m 21s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c4b78c64b06140b39f459de3d7c4a05d : SUCCESS in 19m 11s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80bcf187ed364bb79e15e7bd779eda9a : SUCCESS in 11m 56s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1cb6a9b764d4f0297ddcfd2427e2d41 : SUCCESS in 3m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3bc11e2f7bec23f15e12217c9639fe3b3f66e572","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-11-05 11:13:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27ed39852076752400e058d852a4a8360202f7a6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-11-06 22:20:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c57f949dcbe7af4f356cb58d3a32b6658c814b1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 22:20:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b14acaab03bf14fa3a3e9b98df2fda0f3ec1cf46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 00:12:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5bbcafc8a92e4e16a1c1a4e5d55c5e64 : SUCCESS in 3m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bc6a5c2057b4112bc8903774791823b : SUCCESS in 3m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f345ab126d054eee89e286e0476a520f : SUCCESS in 7m 53s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1538b7c57b149bc8196af75f73b8f0e : SUCCESS in 10m 58s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b3a84c2ba373408bafea8457ba940f5e : SUCCESS in 1h 31m 07s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/03aa0c463bfa4792844c33ff1109c149 : SUCCESS in 1h 49m 01s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f6809ff7409d469386951efb74a84acd : SUCCESS in 1h 26m 17s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b6a6683538474575954e7264ca56ffb4 : SUCCESS in 1h 05m 31s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d3a5628ffda4405d8957f7f590715f57 : SUCCESS in 16m 10s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f49cc7e04a84a09a242afa5da80d49c : SUCCESS in 12m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb7ada7e24ac49c1872e81766a9f2f5a : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"98a1a93ddd59e51df3bd00f26949793e6190175f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 00:12:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"81db800e6bf6e2cd5511cd31ebecd76927f294e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-07 00:13:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bcee2a867754f16b6477f62c8a64267 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8caaed72211265774d472dfc2e9457ee2b4919d6","revisions":{"8caaed72211265774d472dfc2e9457ee2b4919d6":{"kind":"REWORK","_number":1,"created":"2020-11-03 08:14:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/76/761076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/76/761076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/76/761076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/76/761076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/76/761076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/76/761076/1"}}},"commit":{"parents":[{"commit":"eb998ccc4a665b87acd8a23412cddd2ae268496e","subject":"Bump version for the start of Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/eb998ccc4a665b87acd8a23412cddd2ae268496e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-03 08:13:56.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-11-03 08:13:56.000000000","tz":540},"subject":"Add support for the keystone_authtoken/service_type parameter","message":"Add support for the keystone_authtoken/service_type parameter\n\nChange-Id: Ic9df546f304fb0588bd5d18f7979c8df8783e0cb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/8caaed72211265774d472dfc2e9457ee2b4919d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/8caaed72211265774d472dfc2e9457ee2b4919d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
