)]}'
{"id":"openstack%2Fpuppet-placement~782658","triplet_id":"openstack%2Fpuppet-placement~master~I56f95f484b6b30ae7459d4f0e8dfa0cff5cda67a","project":"openstack/puppet-placement","branch":"master","topic":"enforce_new_defaults","hashtags":[],"change_id":"I56f95f484b6b30ae7459d4f0e8dfa0cff5cda67a","subject":"Add support for oslo_policy/enforce_new_defaults","status":"MERGED","created":"2021-03-24 08:06:17.000000000","updated":"2021-03-24 16:37:07.000000000","submitted":"2021-03-24 16:32:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782658-enforce_new_defaults","meta_rev_id":"13ca8112bfcd9512f316e10e09845217e5e7c4ca","_number":782658,"virtual_id_number":782658,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-24 16:32:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-03-24 13:06:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-03-24 14:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-03-24 14:53:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-24 09:45:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-24 13:06:01.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-03-24 14:53:12.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e3f0ee5608c6e16d167397dcd846978a6847796f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-24 08:06:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a07ec676dbf216d789bf7a6b3887f86d2546d5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 09:45:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27fa12d4c4cd4d2b9f11f2fecf828e1d : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8ef33782b72422ebcd4f950cd1f297d : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3bee6836b7784bb08cd102414cb2c3ee : SUCCESS in 9m 42s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/967ef18cb897429fbff974b558353282 : SUCCESS in 6m 23s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f23e947f03ba4ca3817d7ee0ff9c1e00 : SUCCESS in 10m 48s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0a773a62b66448cb8adde42b6aa7cde : FAILURE in 34m 12s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9990ecf9169e4e65b5412a79913288cc : SUCCESS in 54m 53s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d1de33403ee464089013e874d8faac3 : FAILURE in 26m 24s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d494eb8b5684b67819c73ba7684bcd6 : SUCCESS in 42m 39s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/991ea07d436c48cd8050f06b5a8efc33 : SUCCESS in 1h 14m 19s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/24ed02c0f6414b6aa9a5da326f289695 : SUCCESS in 1h 30m 21s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e1d1f8e583284718bed2eebdbf43deaf : SUCCESS in 1h 25m 26s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0cf6e75e6cf4445aa949b8692a25bbeb : SUCCESS in 56m 51s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/ca77339238e94b7da4998527ef693603 : SUCCESS in 18m 08s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/562c96c9c22c4d1d8ab50a44419ad2a3 : SUCCESS in 12m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b672687e3734b7089316adcf884b1ac : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"dd9debe209f7dc774d2d249b2c0bc78347685b95","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-24 13:06:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b1e6b4647f30a50e12c093c19c5370e98e543263","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-24 14:53:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c3f36807645c6057c8d37e1bf398866a228cc1bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 14:54:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68b91e29a58341db036514714244e3579570e1c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 16:32:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cb1f6fd4c2847efabe7560741086656 : SUCCESS in 4m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2225c314ecdb4b43b9c93260b0ead350 : SUCCESS in 4m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/cc267bcb8cd542939521b23eda7ebb42 : SUCCESS in 9m 59s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c477d87ab50f4d0e94199ccda24d9858 : SUCCESS in 7m 35s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0bd529baa1b45079ee981b04e0a3be2 : SUCCESS in 1h 09m 15s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc960d214ab1411ea25e88d759307b36 : SUCCESS in 1h 31m 43s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a04a0d73eb3340a7affef348fd52b4cb : SUCCESS in 1h 32m 39s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8493cb5d74ad4e669ca863e570a8c0a2 : SUCCESS in 1h 00m 01s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/81bb6a205e98437fb76e6466df0fbddd : SUCCESS in 22m 27s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/168d3dd6ba2a4127acc8eef2e1317776 : SUCCESS in 20m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a92ccd806b564217b1bea221aa14265d : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"cc3846246c7d5175b482eda6b93d150fe5a640ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 16:32:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"13ca8112bfcd9512f316e10e09845217e5e7c4ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 16:37:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e7bc702e59b4d32a9512bf678d19950 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3889a08727cf1a385f1012bd411e206647b5488a","revisions":{"3889a08727cf1a385f1012bd411e206647b5488a":{"kind":"REWORK","_number":1,"created":"2021-03-24 08:06:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/58/782658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/58/782658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/58/782658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/58/782658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/58/782658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/58/782658/1"}}},"commit":{"parents":[{"commit":"bff96edd8aa22752ee6c53795b8272d2624970d5","subject":"Merge \"Add support for the oslo_policy/enforce_scope parameter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/bff96edd8aa22752ee6c53795b8272d2624970d5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 08:06:08.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-24 08:06:08.000000000","tz":540},"subject":"Add support for oslo_policy/enforce_new_defaults","message":"Add support for oslo_policy/enforce_new_defaults\n\nDepends-on: https://review.opendev.org/781428\nChange-Id: I56f95f484b6b30ae7459d4f0e8dfa0cff5cda67a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/3889a08727cf1a385f1012bd411e206647b5488a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/3889a08727cf1a385f1012bd411e206647b5488a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
