)]}'
{"id":"openstack%2Fpuppet-placement~785289","triplet_id":"openstack%2Fpuppet-placement~master~I6e340f551d11e3c003f7873a4ccffba262411e52","project":"openstack/puppet-placement","branch":"master","topic":"debian-uwsgi-support","hashtags":[],"change_id":"I6e340f551d11e3c003f7873a4ccffba262411e52","subject":"Add support for placement_api_uwsgi_config in Debian","status":"MERGED","created":"2021-04-07 21:54:14.000000000","updated":"2021-04-14 16:34:22.000000000","submitted":"2021-04-14 16:32:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785289-debian-uwsgi-support","meta_rev_id":"a7e8457c7d79ce77ad2683e966289d9634d67369","_number":785289,"virtual_id_number":785289,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 16:32:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":2,"date":"2021-04-14 14:17:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-04-14 14:34:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-04-14 14:34:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},{"value":0,"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-07 23:44:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-12 06:46:44.000000000","updated_by":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"reviewer":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"state":"REVIEWER"},{"updated":"2021-04-12 06:49:36.000000000","updated_by":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"reviewer":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"state":"REVIEWER"},{"updated":"2021-04-12 06:55:21.000000000","updated_by":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"reviewer":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"state":"REVIEWER"},{"updated":"2021-04-12 07:14:52.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-04-12 07:14:52.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-04-12 07:14:52.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-04-12 07:14:52.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"4d8808d7b60ebea8d2decc1c11d8fc906b034842","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-07 21:54:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9cbc60125cb8d7de0bafc1416f4e66d1bcfcdcbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 23:44:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d0a8db0033f44e7b6aa3d24c15d6c99 : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af0341783bb542dd8a247e4307a104d0 : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/32e2a0a40d3d4e06b0b16df6d8fd0ff9 : SUCCESS in 8m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a8fd7aca14e4fdebf3a32b8f4273203 : SUCCESS in 6m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95902833757b4193a6c480692d427307 : SUCCESS in 5m 58s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfd51ae03dc5438cbac3d9755a10c154 : FAILURE in 31m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1833f4d9121e4f53bb025cd38cc6756b : SUCCESS in 49m 26s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5bad5cd4bab4c87b1ac7c66b68fac3f : FAILURE in 27m 16s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6711ecdf9804a8d865ba5f3a761c62b : SUCCESS in 35m 42s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2d9f5ef9f5840f6b0440fa90de12279 : SUCCESS in 1h 15m 52s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f8b5d352281a496f8cf3db9dd516acdb : SUCCESS in 1h 43m 35s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5fb35a3700841c58246dace95d8b777 : SUCCESS in 1h 29m 10s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/13047b7003504ab9a6a819ba1cacd4e8 : SUCCESS in 57m 10s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/01c1f74418934075af4b26b148631c32 : SUCCESS in 18m 09s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/145f6708fb27441d8fd136e0729bb47e : SUCCESS in 11m 44s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c481ccd803ce42e0b56563cc76cc90ab : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"92c45eaef5668e71665948369cdf7424d45b88ec","author":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"date":"2021-04-12 06:46:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f60b2c4ceac1059d978bf7efab6e4391095aca17","author":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"date":"2021-04-12 06:49:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1b18f19180acb1d57d4f08d8733d66ccfb11ac0c","author":{"_account_id":33357,"name":"Mickael Asseline","display_name":"PAPAMICA","email":"mickael@papamica.com","username":"PAPAMICA"},"date":"2021-04-12 06:55:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"793266c5a9e5c60b865335358f9923485de91653","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-12 16:29:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dfdf0ab8a2efb79e4704e439f9ff66fd01066d85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-12 17:21:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"95380e100ab8a519e1b05eccb72bf9fa9c6070f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 18:36:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fd3abea15b441ceb0d60b925f80a650 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55d99e99b92144b79e059b5d0944bc2f : SUCCESS in 3m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ee09bc09d7034abd88df36caaebdc3a5 : FAILURE in 12m 29s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4fadc5251bf4498b778353d6746525c : FAILURE in 6m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ba1ae3c27204205b8f1ada0aa8e8e53 : FAILURE in 6m 35s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c2f44cf8aa44631a5d4214c1e9c836e : FAILURE in 37m 10s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/102e04e0a2224f59891f80687421e055 : SUCCESS in 46m 23s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ba990ae0f6e4e688c5bf8e7e5af8b47 : FAILURE in 29m 56s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a21d733eff84c3383a19bf9c4b0dbcd : SUCCESS in 34m 01s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/61f712628f2442e2a26d5a713aef934b : SUCCESS in 1h 07m 02s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/54e03900b40d48af9156a412d876e9a2 : SUCCESS in 1h 11m 10s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/db5847dda8f94c1f803c8d8b83dce872 : SUCCESS in 1h 13m 22s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5e1fe8e87eb4f5cbddf17ddce3e999e : SUCCESS in 54m 09s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1bb02e42fed48188654c871df8a4688 : SUCCESS in 17m 08s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17db88b177c344a3b6653de5f9b1cd0c : SUCCESS in 11m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65b5d0d32d27486da74c4f7172542c90 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":3},{"id":"d141016f7afdcddf5c58b89f6f9939aff7a1311d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-12 19:03:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"571358a87a533e1ab7f5fc01c272e151f41d5f0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-12 19:05:30.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ae62d835f6bf6703779ad752c363b439ec117914","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-04-12 19:27:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"528f319c4679a7fdaeebf973159dac6d71645db3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 20:58:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2f6f0be2c3849d189360bef1b250051 : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2533c66b3c54cb2ba23a18790526e01 : SUCCESS in 4m 30s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1cc1dd85457143899b5d77cdcfc67b8e : SUCCESS in 12m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3944f82f4494193884fe145c866af93 : SUCCESS in 9m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7bc9d8ccf9343a3bf1a24eff5325962 : SUCCESS in 6m 57s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cd8106f53df4db495390e0df549c381 : FAILURE in 41m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/036c1c3c544046ae9c186cd633344edf : SUCCESS in 1h 02m 14s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12864891eed7452ba18a0b1f134ba6d8 : FAILURE in 31m 08s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6adffb5530e40bab8c0fbd550ca98e2 : SUCCESS in 40m 21s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/950dcdef02c2457885af7764c69ac59e : SUCCESS in 1h 10m 34s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4551a31389254e79938137e5dd71a760 : SUCCESS in 1h 22m 10s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/54680b17c9a1488c9cbcf34fa6414f7b : SUCCESS in 1h 26m 29s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/dbc6d357c2c04c238747d1d883923ead : SUCCESS in 54m 14s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/f61f722c83aa4b4384c9f60c2b919e99 : SUCCESS in 17m 46s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f17eef415ea54c0a8fd42142250ace47 : SUCCESS in 13m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46107f8376fc4e7683c853c4e9090e67 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":6},{"id":"89b9d32d99a94ec4ca608b0de6758a25f4b93b29","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-14 14:17:56.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"2fc0844d3c5d419530690ace3f813143b4a314bd","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-04-14 14:34:59.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"b6583d1a864378eeb1da31bebfc26bf617a2fef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 14:35:11.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e1e374d6f2ee7ca73ab9263bb4892b3f9c7e46bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:32:07.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40590c4c06ea414585c6fe9382f07697 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba8c70c62dd148ec9271f32aaeb375b9 : SUCCESS in 3m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ed66118227c4f4aa2c5803ea6f6f31d : SUCCESS in 8m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2ee25479ab049f599160ba7faf9fc54 : SUCCESS in 6m 24s\n- puppet-openstack-integration-6-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0075d4fca33a435aa2060a70f1a70380 : SUCCESS in 1h 18m 32s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6492356e3aa44108d7fe00e3f02b8fa : SUCCESS in 1h 51m 50s\n- puppet-openstack-integration-6-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f45f3722722148718db35f135f121fd7 : SUCCESS in 1h 28m 22s\n- puppet-openstack-integration-6-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3cfe4cbaa69a45d6b5ec653b72e21037 : SUCCESS in 59m 19s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/5334bbeedabc49b78e1f6ad7ab6c5a7f : SUCCESS in 20m 52s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f8fdff4048a4c82bcceda09e87e968f : SUCCESS in 14m 46s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/153b13ce840945a2bd7cd25a8326939f : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":6},{"id":"3b670d7f580c4cc4785dfa1b28d38e1e79e63ac5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:32:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"a7e8457c7d79ce77ad2683e966289d9634d67369","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:34:22.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/195bd643bd594cb5a811f506c80788d5 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"f5c124170afffcf9aac52b6429de96c08a615b93","revisions":{"60e0d34c489f2327717defa1f124993c5b427758":{"kind":"REWORK","_number":1,"created":"2021-04-07 21:54:14.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/89/785289/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/89/785289/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/1"}}},"commit":{"parents":[{"commit":"e9d25d00d90a392060037ae2686b2a5c3ee897f0","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/e9d25d00d90a392060037ae2686b2a5c3ee897f0"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 21:53:35.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 21:53:35.000000000","tz":120},"subject":"Add support for placement_api_uwsgi_config in Debian","message":"Add support for placement_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Placement API. Therefore, this patch adds a new\nplacement_api_uwsgi_config provider as well as a new\nplacement::wsgi::uwsgi class.\n\nChange-Id: I6e340f551d11e3c003f7873a4ccffba262411e52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/60e0d34c489f2327717defa1f124993c5b427758"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/60e0d34c489f2327717defa1f124993c5b427758"}]},"branch":"refs/heads/master"},"5f3e0b4f312beaa3006f00818cfc8b37e3430e86":{"kind":"REWORK","_number":2,"created":"2021-04-12 16:29:50.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/89/785289/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/89/785289/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/2"}}},"commit":{"parents":[{"commit":"e9d25d00d90a392060037ae2686b2a5c3ee897f0","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/e9d25d00d90a392060037ae2686b2a5c3ee897f0"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 21:53:35.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-12 16:29:36.000000000","tz":120},"subject":"Add support for placement_api_uwsgi_config in Debian","message":"Add support for placement_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Placement API. Therefore, this patch adds a new\nplacement_api_uwsgi_config provider as well as a new\nplacement::wsgi::uwsgi class.\n\nChange-Id: I6e340f551d11e3c003f7873a4ccffba262411e52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/5f3e0b4f312beaa3006f00818cfc8b37e3430e86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/5f3e0b4f312beaa3006f00818cfc8b37e3430e86"}]},"branch":"refs/heads/master"},"21c5f17730e4b2c337dc65218f6901bfc2ee9ae5":{"kind":"REWORK","_number":3,"created":"2021-04-12 17:21:11.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/89/785289/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/89/785289/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/3"}}},"commit":{"parents":[{"commit":"e9d25d00d90a392060037ae2686b2a5c3ee897f0","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/e9d25d00d90a392060037ae2686b2a5c3ee897f0"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 21:53:35.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-12 17:20:59.000000000","tz":120},"subject":"Add support for placement_api_uwsgi_config in Debian","message":"Add support for placement_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Placement API. Therefore, this patch adds a new\nplacement_api_uwsgi_config provider as well as a new\nplacement::wsgi::uwsgi class.\n\nChange-Id: I6e340f551d11e3c003f7873a4ccffba262411e52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/21c5f17730e4b2c337dc65218f6901bfc2ee9ae5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/21c5f17730e4b2c337dc65218f6901bfc2ee9ae5"}]},"branch":"refs/heads/master"},"7142f88971a776bbc94ee56d534a8f004e466e40":{"kind":"REWORK","_number":4,"created":"2021-04-12 19:03:01.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/89/785289/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/89/785289/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/4"}}},"commit":{"parents":[{"commit":"e9d25d00d90a392060037ae2686b2a5c3ee897f0","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/e9d25d00d90a392060037ae2686b2a5c3ee897f0"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 21:53:35.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-12 19:02:50.000000000","tz":120},"subject":"Add support for placement_api_uwsgi_config in Debian","message":"Add support for placement_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Placement API. Therefore, this patch adds a new\nplacement_api_uwsgi_config provider as well as a new\nplacement::wsgi::uwsgi class.\n\nChange-Id: I6e340f551d11e3c003f7873a4ccffba262411e52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/7142f88971a776bbc94ee56d534a8f004e466e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/7142f88971a776bbc94ee56d534a8f004e466e40"}]},"branch":"refs/heads/master"},"8fe3478dad8cc6827df8ec60d326f44b7251942b":{"kind":"REWORK","_number":5,"created":"2021-04-12 19:05:30.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/89/785289/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/89/785289/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/5"}}},"commit":{"parents":[{"commit":"e9d25d00d90a392060037ae2686b2a5c3ee897f0","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/e9d25d00d90a392060037ae2686b2a5c3ee897f0"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 21:53:35.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-12 19:05:19.000000000","tz":120},"subject":"Add support for placement_api_uwsgi_config in Debian","message":"Add support for placement_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Placement API. Therefore, this patch adds a new\nplacement_api_uwsgi_config provider as well as a new\nplacement::wsgi::uwsgi class.\n\nChange-Id: I6e340f551d11e3c003f7873a4ccffba262411e52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/8fe3478dad8cc6827df8ec60d326f44b7251942b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/8fe3478dad8cc6827df8ec60d326f44b7251942b"}]},"branch":"refs/heads/master"},"f5c124170afffcf9aac52b6429de96c08a615b93":{"kind":"REWORK","_number":6,"created":"2021-04-12 19:27:21.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/89/785289/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/89/785289/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/89/785289/6"}}},"commit":{"parents":[{"commit":"e9d25d00d90a392060037ae2686b2a5c3ee897f0","subject":"Merge \"Allow to configure policy_dirs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/e9d25d00d90a392060037ae2686b2a5c3ee897f0"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-07 21:53:35.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-04-12 19:27:10.000000000","tz":120},"subject":"Add support for placement_api_uwsgi_config in Debian","message":"Add support for placement_api_uwsgi_config in Debian\n\nThis patch is adding the configuration of the number of workers,\nthreads, and the size of the listen queue in Debian, which uses\nuwsgi to run Placement API. Therefore, this patch adds a new\nplacement_api_uwsgi_config provider as well as a new\nplacement::wsgi::uwsgi class.\n\nChange-Id: I6e340f551d11e3c003f7873a4ccffba262411e52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/f5c124170afffcf9aac52b6429de96c08a615b93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/f5c124170afffcf9aac52b6429de96c08a615b93"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
