)]}'
{"id":"openstack%2Fpuppet-placement~808881","triplet_id":"openstack%2Fpuppet-placement~master~Ic468378d42df5b07bb053f8e318fd383b4964874","project":"openstack/puppet-placement","branch":"master","topic":"bug/1943212","hashtags":[],"change_id":"Ic468378d42df5b07bb053f8e318fd383b4964874","subject":"Add watch_log_file option","status":"MERGED","created":"2021-09-14 08:02:55.000000000","updated":"2021-09-15 03:14:21.000000000","submitted":"2021-09-15 03:13:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"808881-bug/1943212","meta_rev_id":"a7c937ef2c52b64af35367f387a46375998466e7","_number":808881,"virtual_id_number":808881,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-15 03:13:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-09-15 01:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-09-14 15:12:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-09-15 01:11:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-14 10:05:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-14 15:12:00.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"6ae6381943661ae7066cd286ea278a90debc720e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-14 08:02:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31874c3c62f9195fdfb6502565072751971f2ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 10:05:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb18eaf3f97b499e92ba6519adc160e4 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8ec9a13a55248d09fdb3c857a948ba4 : SUCCESS in 6m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/669a5b4dc3744ef095a745bf904737be : SUCCESS in 4m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/150d75cd0a554a178ffbc88b2d31dff0 : SUCCESS in 11m 25s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ca3fd3dd5064eaaaba8d32e82a2bba3 : SUCCESS in 12m 57s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa62e83a448049d7841dbf4b698c1a21 : SUCCESS in 9m 20s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d26cbcf945f4c16b424411d95849227 : SUCCESS in 8m 31s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b3aaff720304adcaaa775b28758bbbd : SUCCESS in 11m 30s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fecbfe351dd4aa592bebbc3589e2457 : SUCCESS in 47m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/244153b83e084d63bdb03a1ce213ce9b : SUCCESS in 57m 28s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04490956909d484c89e0c00196e80fd1 : FAILURE in 44m 56s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fed80102d934cbc953ad14d16223079 : SUCCESS in 40m 54s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c3e068955f2a40b8a7f879c2b9da64e1 : SUCCESS in 25m 13s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c26f199c1ca74261b1fa080e409c07b2 : SUCCESS in 1h 11m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ce3a085e4f04a76b19bfedd38c900c1 : SUCCESS in 1h 24m 42s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ecd0527ea348491ab0ce20d6acf9ead7 : SUCCESS in 1h 33m 34s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fd4821041353465db92af30f0b4f4a8c : SUCCESS in 1h 01m 18s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4988cb95cb1848cb83e4075ee428f752 : SUCCESS in 17m 15s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1e77a672efb74126a34ec852be5abe55 : SUCCESS in 21m 27s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d252c512039d4fdea8e834987f62bd9a : SUCCESS in 15m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4023d0d27ee440898d89eca925010cae : SUCCESS in 16m 19s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38f1ef1ab5f94340be4c5734f31934f4 : SUCCESS in 7m 43s\n- puppet-placement-tripleo-standalone https://zuul.opendev.org/t/openstack/build/06ea6d9c03bd408f9a49328c5ec23297 : SUCCESS in 1h 58m 20s","accounts_in_message":[],"_revision_number":1},{"id":"6f291ada510439e309a165d16fdb897da80d5faf","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-14 15:12:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0a1fdf3a48eda51ffa07d9d3285f1b24ff1ab0f3","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-09-15 01:11:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"98fc896a4064ca801a6ead168bace88a2d7e023e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 01:12:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"09c00f57ebe15e5419b48909cf82306a82960c87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 03:13:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa664f5359124443b44e1a62182a1579 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ff9bbf0439044de9ff22af982a883ad : SUCCESS in 6m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db9c62afa9e04dc9a62fdbe8635c0519 : SUCCESS in 5m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/af9d34ef2af441209253894c2e010416 : SUCCESS in 10m 23s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/764a1e2d19d6416c9a807216ffb10c92 : SUCCESS in 21m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/592f0141287a46368c24e0e1b9a781e1 : SUCCESS in 10m 56s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddc8ae63967e4ae790f663a58c881e22 : SUCCESS in 8m 37s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/566624520677443595ce2f910a8de5d7 : SUCCESS in 1h 23m 13s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f9f7912790340478032a6224dd91968 : SUCCESS in 1h 30m 06s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/377a4a1a270f4dd6a09c6780ce48af4a : SUCCESS in 1h 36m 45s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d16bdf8583bb46fb98bc94513b42778f : SUCCESS in 1h 12m 51s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/8337d8e2b51647d29cfdd44b6496c223 : SUCCESS in 18m 44s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbc8af7024a34f58a372102dec605dfd : SUCCESS in 13m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41c7aa79928b4ad4b89d0a86a6f72fc5 : SUCCESS in 3m 59s\n- puppet-placement-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6e893365b32c47e6a7d44efcffa9e7c8 : SUCCESS in 1h 55m 18s","accounts_in_message":[],"_revision_number":1},{"id":"4b358dafb3c2b60594f1072506d693a5632a8743","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 03:13:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a7c937ef2c52b64af35367f387a46375998466e7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 03:14:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a90654629c8b40478a22722518a82773 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42770dd45e4f4d5452bd7ed55b9fd6f97c8400d8","revisions":{"42770dd45e4f4d5452bd7ed55b9fd6f97c8400d8":{"kind":"REWORK","_number":1,"created":"2021-09-14 08:02:55.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/81/808881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/81/808881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/81/808881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/81/808881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/81/808881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/81/808881/1"}}},"commit":{"parents":[{"commit":"90052058ad6f37506f90cc23dbf16788e33c0ceb","subject":"Merge \"Allow purging policy files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/90052058ad6f37506f90cc23dbf16788e33c0ceb"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-09-14 08:02:09.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2021-09-14 08:02:09.000000000","tz":480},"subject":"Add watch_log_file option","message":"Add watch_log_file option\n\nAdd support for Using logging handler designed to watch file system.\n\nChange-Id: Ic468378d42df5b07bb053f8e318fd383b4964874\nCloses-Bug: #1943212\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/42770dd45e4f4d5452bd7ed55b9fd6f97c8400d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/42770dd45e4f4d5452bd7ed55b9fd6f97c8400d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
