)]}'
{"id":"openstack%2Fpuppet-placement~811061","triplet_id":"openstack%2Fpuppet-placement~master~I4f44681800d912e171b69239bd0c85bf231ef191","project":"openstack/puppet-placement","branch":"master","topic":"puppet-service-desc","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-23 18:25:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4f44681800d912e171b69239bd0c85bf231ef191","subject":"Use consistent format for service description","status":"MERGED","created":"2021-09-27 03:19:24.000000000","updated":"2021-11-23 18:25:37.000000000","submitted":"2021-11-23 18:25:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"811061-puppet-service-desc","meta_rev_id":"06623fe6acf7a69f841d0e06d2893761e0448e01","_number":811061,"virtual_id_number":811061,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-23 18:25:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-11-18 22:26:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-11-23 16:38:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-11-23 16:38:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-27 05:18:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-18 22:26:26.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-11-23 16:38:50.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"3530b862cee71331162d29b11b30e9c7b854e372","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-27 03:19:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b461dad0fb63bf8c13caf5523da75ebb4410764","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-27 03:19:33.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"051715e41791e0a5e0223971bce4cb9265f01890","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 05:18:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acba680934924b66a6fc2ac02f677b69 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08674d8ac3bb4dbd89ee269e1fa92c1f : SUCCESS in 3m 47s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c08d3a17f91644a8ae1f573a592f56f0 : SUCCESS in 3m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e95e94fbff414079a4eb1b6c21bf1020 : SUCCESS in 13m 17s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/458d2d8855324fec91e30f9e73da82e4 : SUCCESS in 10m 12s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08208138140f4585b3fb63196918a5b5 : SUCCESS in 7m 24s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1dffe82adfa46799ef184e602135196 : SUCCESS in 10m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ba6bfe2e89d4d998931b87dd1e3df0b : SUCCESS in 7m 40s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d985e6b80fe47cc9ed3ef6b10fa12ae : SUCCESS in 49m 43s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7b6b5cc2e5843cca536df90e33be7c0 : SUCCESS in 49m 08s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27fbee646afc44cc8cd5ee162eb5cf4a : FAILURE in 31m 04s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d328278c00aa4653be11596362c67c42 : SUCCESS in 35m 35s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0a1d919b9a0f48dba55df49d44c9d55a : SUCCESS in 38m 16s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/807e5212a04a423888921d18c77c84e4 : SUCCESS in 1h 27m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/33dd1477595f4761941a9ffc14ab4313 : SUCCESS in 1h 40m 23s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4b068c39eb644389a927e152bbc5eaa3 : SUCCESS in 1h 22m 39s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9bf907fec0a4bbe891bc4e10bf7aaff : FAILURE in 1h 01m 45s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6719e17334df4489853720161d4f3dbc : SUCCESS in 13m 40s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3070d80800f444982fb70578193cc3f : SUCCESS in 20m 05s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e9fe199fb604abc9d42dca19b020de0 : SUCCESS in 14m 20s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d682554d72d443c7b8e365c8d75ac662 : SUCCESS in 25m 56s (non-voting)\n- puppet-placement-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fc193f2e01704be8b74416b809492960 : SUCCESS in 1h 54m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d74a39cb70b83d68455cce256d9cf4b5ca007c77","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-29 04:09:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5142a4f444d362f195cab57de3992f7212a19873","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-29 05:57:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebb7beebbeee444dbc994e99fff5743a : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb4c1593ac9e4bf5902f482251d02d1e : SUCCESS in 3m 57s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/030beaf694664766aa35b09886188bab : SUCCESS in 4m 12s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a5d21b642e2645e7a7e83fc65df9c03d : SUCCESS in 9m 46s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ec162b630a6246438adf4001ca1f3f9b : SUCCESS in 11m 07s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8da42c239b8649659e4087be104305e9 : SUCCESS in 8m 03s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dfa7f00316b40649a97564426d0ba22 : SUCCESS in 8m 47s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27923b85bde444479f3ea1922d8fed5d : SUCCESS in 8m 08s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbbda9afc0ad4f11bb390091bea73b98 : FAILURE in 7m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c6af17ccfd1447eb06b5dabb2ebbdd0 : FAILURE in 7m 03s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87de5f64317b4cd482a3cfcbe08fef7d : FAILURE in 10m 03s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95875092dae5415ba88b65e947a9db84 : FAILURE in 7m 50s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/43ce0949b08c4fcca0ddbc642c892d50 : SUCCESS in 25m 12s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d9d7a761db714f808a2b76a944278bc5 : SUCCESS in 1h 13m 41s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1875e669cd93418995db8cd67f46b1ff : SUCCESS in 1h 19m 46s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d771f6a92cf4ad8ad9f86f5f90a86d6 : SUCCESS in 1h 16m 16s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/005bfb79d3fc48df8971c1182c28d2b3 : SUCCESS in 1h 03m 39s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/852dac57b77148a3b241129bc01954ba : SUCCESS in 16m 21s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/886f8244a94441dfb18aa37f1728c7e2 : SUCCESS in 20m 15s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffb01c2c051d4e00a3f5cc91591d6e87 : SUCCESS in 15m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8216b5b458a54b7685869f0734cc36b7 : SUCCESS in 16m 40s (non-voting)\n- puppet-placement-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b6dc88e6109f4d538c1a6b1a24ada155 : SUCCESS in 1h 44m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a6cf6955c4f31d8ed651fd5c60b7459126ccf3c0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-18 14:31:10.000000000","message":"Topic auth-ut removed","accounts_in_message":[],"_revision_number":1},{"id":"eb5cad38093a9cc003bcfe7db6ecbc666700b267","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-18 14:31:13.000000000","message":"Topic set to service-name","accounts_in_message":[],"_revision_number":1},{"id":"c4303b2ca983cecefb9a0ebfa27e74a988f4b201","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-18 14:31:17.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"9e14e4a97fbaa1eb4ae6cb3b8bccc586fe09a018","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-18 14:32:07.000000000","message":"Topic service-name removed","accounts_in_message":[],"_revision_number":2},{"id":"ac486b6420ec834119bfe6803e92b63309d011be","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-18 14:32:09.000000000","message":"Topic set to puppet-service-desc","accounts_in_message":[],"_revision_number":2},{"id":"58cec3cb99201b3c646f161e48bd1574e96bd211","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-18 14:32:21.000000000","message":"Topic puppet-service-desc removed","accounts_in_message":[],"_revision_number":2},{"id":"bdc1edc1e43aa0de633eb44a20e16f4274c60274","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-18 14:32:26.000000000","message":"Topic set to puppet-service-desc","accounts_in_message":[],"_revision_number":2},{"id":"b44dee1283f977a67c6f5094c98fa088b0332b46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 17:04:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f8b88c74bdb4aaba8116759f373542d : SUCCESS in 3m 57s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bc23a02be2249f68133050efe055952 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89490b492f834c078eb8f1c1d28c73d2 : SUCCESS in 3m 58s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/abb6a55f159242279cadd1947002a6a6 : SUCCESS in 14m 55s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3b25c82ec1804f01899ae586e45a610a : SUCCESS in 11m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02eee301700a4fa6a73ce564691c779d : SUCCESS in 11m 10s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6a4ff3c07b84f5a91caab63b0708b9f : SUCCESS in 11m 50s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/552d82fef2744dd2bde90160b0902f64 : SUCCESS in 11m 28s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60f120465792410ea60400b4c23ab4de : FAILURE in 1h 05m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55a65fca264b44f583456f7b444aa2d2 : FAILURE in 1h 16m 32s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/496a80602bdb46339710aaeba4eaab81 : FAILURE in 33m 36s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1753787b61734920a80f6aa6d1217374 : FAILURE in 54m 13s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d8a297b621b4f3e927c3daa436e3f77 : SUCCESS in 23m 47s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/242021f997f64b67a124b69bb9669b8a : SUCCESS in 1h 33m 17s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/52633913770545a98b8a2dd056d163ae : SUCCESS in 1h 40m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0602c95b181544949f69843812985ff1 : SUCCESS in 1h 21m 04s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b6f9f0a89baf4b8b8b65b3030aaa3611 : SUCCESS in 58m 18s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3edd8d5bc4dd47c3b8410331fee13135 : SUCCESS in 13m 27s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ef7889437a9b402a96a2ac9ea66da315 : SUCCESS in 21m 45s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12777516987942fa9b4d07dca7b5d2e3 : SUCCESS in 13m 17s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/beaaf35dcc2343158626012716aafd93 : SUCCESS in 15m 37s (non-voting)\n- puppet-placement-tripleo-standalone https://zuul.opendev.org/t/openstack/build/341ff70dd9314bf5bed254734daab957 : SUCCESS in 2h 23m 25s","accounts_in_message":[],"_revision_number":2},{"id":"ab94ca9a88cca760b450c04570865a6277c6c8e8","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-18 22:26:26.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1ac844afad0a7a38d161a17f9538147cc28b86a0","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-23 16:38:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"299d064611e3b4ed8b37223749a7a43d20085b4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 16:43:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6a72e9e1f30ecdef4a1a808a13017ddbada738f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 18:25:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/460af992667843888292aeb97c4d21b2 : SUCCESS in 6m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bc90e955ce8449a806dba4ddbf73c63 : SUCCESS in 5m 15s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd4232447d0a4fb29d2fbbf34e26b9aa : SUCCESS in 4m 15s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/398420aabd5548109efd43837a665d5f : SUCCESS in 16m 37s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9dcb13482ca418a8aed76164834e68a : SUCCESS in 12m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc0dcbbe40eb454a9c7dca46fa7930f6 : SUCCESS in 9m 18s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b88783ae3fa4c33a2e56baf0e5b84a7 : SUCCESS in 12m 45s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f8b9d73ac4c44afb23193dcb94bc636 : SUCCESS in 1h 21m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0767af09f20d43c2a8047fedd35af3f6 : SUCCESS in 1h 29m 04s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6ca445a8a3b34e58b409f5174257487d : SUCCESS in 1h 25m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/06fde367737d4b2185c64c8d291a824f : SUCCESS in 1h 11m 59s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c280d4430c9b4c249894743c672c8665 : SUCCESS in 22m 31s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21827c87569e46d2b316899175e80195 : SUCCESS in 21m 37s (non-voting)\n- puppet-placement-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7c106a9962b84defa876f6cc08702771 : SUCCESS in 1h 34m 27s","accounts_in_message":[],"_revision_number":2},{"id":"06623fe6acf7a69f841d0e06d2893761e0448e01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 18:25:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0409934acec3735b8ca4e413a85a1e01a5c9a603","revisions":{"62dd6f0acf30cc641821b77db971d99669b20879":{"kind":"REWORK","_number":1,"created":"2021-09-27 03:19:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/811061/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/61/811061/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/61/811061/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/61/811061/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/61/811061/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/61/811061/1"}}},"commit":{"parents":[{"commit":"f00edeccacdd4b4f41733dcd7b24c290a06e7d4e","subject":"Do not test detail of keystone::resource::service_identity","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/f00edeccacdd4b4f41733dcd7b24c290a06e7d4e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-27 03:19:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-27 03:19:06.000000000","tz":540},"subject":"Use consistent format for service description","message":"Use consistent format for service description\n\nChange-Id: I4f44681800d912e171b69239bd0c85bf231ef191\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/62dd6f0acf30cc641821b77db971d99669b20879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/62dd6f0acf30cc641821b77db971d99669b20879"}]},"branch":"refs/heads/master"},"0409934acec3735b8ca4e413a85a1e01a5c9a603":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-11-18 14:31:17.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/811061/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/61/811061/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/61/811061/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/61/811061/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/61/811061/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/61/811061/2"}}},"commit":{"parents":[{"commit":"467c0b0a4da61826d72e3af8f781b010ad519feb","subject":"update file name for apache config on Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/467c0b0a4da61826d72e3af8f781b010ad519feb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-27 03:19:06.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-18 14:31:17.000000000","tz":0},"subject":"Use consistent format for service description","message":"Use consistent format for service description\n\nChange-Id: I4f44681800d912e171b69239bd0c85bf231ef191\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/0409934acec3735b8ca4e413a85a1e01a5c9a603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/0409934acec3735b8ca4e413a85a1e01a5c9a603"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
