)]}'
{"id":"openstack%2Fpuppet-placement~816235","triplet_id":"openstack%2Fpuppet-placement~stable%2Ftrain~If39b4cabadc2d7d5c33ef7ac2b56d6adfdcdf456","project":"openstack/puppet-placement","branch":"stable/train","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-02 23:13:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If39b4cabadc2d7d5c33ef7ac2b56d6adfdcdf456","subject":"Tag the httpd service only when it is used","status":"MERGED","created":"2021-11-02 00:10:00.000000000","updated":"2021-11-02 23:13:43.000000000","submitted":"2021-11-02 23:13:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816235","cherry_pick_of_change":816234,"cherry_pick_of_patch_set":1,"meta_rev_id":"b6704cb1bafb56f42be62322795e5c7ce46e3258","_number":816235,"virtual_id_number":816235,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-02 23:13:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-02 15:45:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-11-02 21:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-11-02 21:45:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-02 01:15:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-02 15:45:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-11-02 21:45:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"b563e7a86ed2d22b1c761b753dc331a0b4b3c487","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 00:10:00.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"45e5e7293c2ee3d3eadbe8a6974e7ec7ed23f72a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 01:15:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52d29fbdd6b0471fba056524f51390a3 : SUCCESS in 5m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d5dc4ff57fb4464aadf068618e56fdbe : SUCCESS in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b55ff3ce47f48afa2579cc6dc06600a : SUCCESS in 5m 31s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1ebdbdb5fafe4490b256d10c3b8bec5c : SUCCESS in 20m 41s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ee9effe06a124af0b2318f44f7fc6d28 : SUCCESS in 11m 04s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/28f2f279ae934d31b6a170cc68b425d8 : SUCCESS in 10m 45s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6ee9706d5f444cfb023e28566a088cd : SUCCESS in 12m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0442196b4f5445e8b99b6a927d25ed6a : SUCCESS in 9m 34s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/4b545894c77e454985d85edc9e096c18 : SUCCESS in 53m 24s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/bf750389595f4d9aa5f1af2e6b493053 : SUCCESS in 1h 00m 10s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/7f4dea0ffc6e42078b12762026dbad60 : SUCCESS in 57m 52s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/8975e9d665aa40b69d7d94e4b952686e : SUCCESS in 47m 04s\n- puppet-openstack-integration-5-scenario001-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/543db89b39244244add169ad9363a289 : SUCCESS in 48m 40s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0fd07fc27a54514a908e366376e2e39 : SUCCESS in 42m 54s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/835c80458f724adcad987f571cbe6b19 : SUCCESS in 34m 58s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-ubuntu-bionic-mimic https://zuul.opendev.org/t/openstack/build/8a7c11f7a9574fd0b7053f01d9daaa9b : SUCCESS in 32m 08s (non-voting)\n- puppet-openstack-integration-5-scenario001-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/af8cb35ff8f74f4b93b0a645ff59d7ab : SUCCESS in 49m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ac215b4f5a4c478a859bf331feed67bc : SUCCESS in 1h 00m 23s (non-voting)\n- puppet-openstack-integration-5-scenario003-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/359ab0dff3a041bba8f75fe6600ebbc1 : SUCCESS in 53m 51s (non-voting)\n- puppet-openstack-integration-5-scenario004-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aaa06ad5e52644b7b54c8efd76fdb39d : SUCCESS in 43m 49s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9001e7d4153849ed95e93b1029f503d7 : SUCCESS in 10m 17s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c8795adb4a54427846f9558f0a2dbe8 : FAILURE in 9m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d8e12363a7bb6c3ad91f40819335c04aa8961fa","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-02 15:45:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"69e81377d51aad27fc6f329261f6700c8415a845","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-02 21:45:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15c4b9c43ca51b4e06895bd4af0be3b334a63e13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 21:46:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"820ecdc93f1d92e247089ec5ad91efe50806a1a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 23:13:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/27dd5fe9235e49d39328fe5c7b3a551a : SUCCESS in 5m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a61a4dac07a04e9a8bce2329c57c9f29 : SUCCESS in 5m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30c25264147a4651946934174b1a2171 : SUCCESS in 5m 19s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/bfda8534bdbf43c4bc6f100173cc5dbd : SUCCESS in 9m 53s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b3286e12f964a5481be68c80d67b1c1 : SUCCESS in 9m 19s\n- puppet-openstack-integration-5-scenario001-tempest-centos-7-luminous https://zuul.opendev.org/t/openstack/build/3f9a93342f9443ddbed4b8112924273a : SUCCESS in 50m 19s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/d4ed64edd872478eadef5bef279cdd11 : SUCCESS in 1h 15m 50s\n- puppet-openstack-integration-5-scenario003-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/521f13223a71471582daadee091407ca : SUCCESS in 58m 20s\n- puppet-openstack-integration-5-scenario004-tempest-centos-7-nautilus https://zuul.opendev.org/t/openstack/build/956fe763bfbc42f9b14ea62b67087510 : SUCCESS in 58m 30s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9a58f867f95b49d3afa2f8d8821fe680 : SUCCESS in 12m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b6704cb1bafb56f42be62322795e5c7ce46e3258","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 23:13:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"65f9bf4ad6092cda28f4544d0de8710ce06ee8f8","revisions":{"65f9bf4ad6092cda28f4544d0de8710ce06ee8f8":{"kind":"REWORK","_number":1,"created":"2021-11-02 00:10:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/816235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/35/816235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/35/816235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/35/816235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/35/816235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/35/816235/1"}}},"commit":{"parents":[{"commit":"8fe110ecea431a9e5380e1295989158c43bdf2e2","subject":"Prepare the final stable/train release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/8fe110ecea431a9e5380e1295989158c43bdf2e2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-01 08:31:04.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 00:10:00.000000000","tz":0},"subject":"Tag the httpd service only when it is used","message":"Tag the httpd service only when it is used\n\nCurrently the placement-service tag is always added to the httpd\nservice, but it should not if an independent service is used to run\nplacement-api.\n\nChange-Id: If39b4cabadc2d7d5c33ef7ac2b56d6adfdcdf456\n(cherry picked from commit 8475c1afed91ded3e393c1e2f41daf8c98b95bde)\n(cherry picked from commit 14ee3c820b89597dd95d3abd2c335ac7a4075d03)\n(cherry picked from commit 568b61b585f710a778e34a83a3c6a21b17c77b55)\n(cherry picked from commit 314bbbcc72320fc71fd6092e720d8b7a3ae819f8)\n(cherry picked from commit 9e4cf669ba037df9fb8e04b11c40bd560381882e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/65f9bf4ad6092cda28f4544d0de8710ce06ee8f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/65f9bf4ad6092cda28f4544d0de8710ce06ee8f8"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
