)]}'
{"id":"openstack%2Fpuppet-placement~819535","triplet_id":"openstack%2Fpuppet-placement~master~Ie9c8fa687ca57d7bc49d6f35941efbc94a3f5344","project":"openstack/puppet-placement","branch":"master","topic":"policy-deps","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-29 13:52:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie9c8fa687ca57d7bc49d6f35941efbc94a3f5344","subject":"Fix dependencies related to openstacklib::policy","status":"MERGED","created":"2021-11-29 00:57:54.000000000","updated":"2021-11-29 13:52:21.000000000","submitted":"2021-11-29 13:52:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819535-policy-deps","meta_rev_id":"c3ae05ca27fce4590e84ce5a887f9bc17fe2de36","_number":819535,"virtual_id_number":819535,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-29 13:52:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-11-29 11:41:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-11-29 08:18:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-11-29 11:41:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-29 02:51:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-29 08:18:49.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-11-29 11:41:42.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"0fd15fb110bf8f5311dc11b9d3e2f36269c80695","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 00:57:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4eceb5061979c860125117d3c1706acf979e030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 02:51:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01aa3db06afd47e8af03b275701947e0 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3e8820157e84497bc8ecca5dc00162e : SUCCESS in 3m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01d61d0de95d42cdaa1e387ba7865340 : SUCCESS in 4m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/98dd3de1cc2448eea490d4014136791f : SUCCESS in 10m 35s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/39d64204c4a54731ab50f31ff3ba07ae : SUCCESS in 9m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cbe311ae93843928888897153cf2db4 : SUCCESS in 8m 55s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/601189d56cf34d98b22d4ed50f0bf5ac : SUCCESS in 8m 49s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96b2f9fdd21947a4a175e884b59b4bc9 : SUCCESS in 8m 13s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/beee491afef24c998e8cefc66871b8f7 : FAILURE in 29m 39s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71172802841a47c9b587f3b340ffc478 : FAILURE in 1h 16m 58s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebbc09c9b8cd4ea6a8579a6002e483bb : FAILURE in 28m 53s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a01a5a0a1dcb4f94ba7367e1d90e7c7e : FAILURE in 59m 49s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8683207d94fb4995acca31ad9db9bb41 : SUCCESS in 20m 35s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/87895467f0f04c3f971439dbb77bd2c1 : SUCCESS in 1h 13m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b920c0768fe44efab46e298d90b0e031 : SUCCESS in 1h 30m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4798d935ae08424088449c2b44655da1 : SUCCESS in 1h 22m 59s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/589ce1d454d74caa8fe5f073fb57fe9f : SUCCESS in 1h 25m 20s\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2a9fee0f5d24406db31675813cc6097e : SUCCESS in 16m 53s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/eeb4e28bfbaa4192a7e70a8a29dfc8ee : SUCCESS in 52m 57s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca60d1e297654b8cbd6292d4bbb1b76e : SUCCESS in 51m 25s (non-voting)\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/70be73ab3438481285d87ed5b920c655 : SUCCESS in 58m 11s (non-voting)\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/679054c97cf14f98bd5cc6eb2f2ff9be : SUCCESS in 49m 31s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/016793315b25473cbe5e9bae5dc71c2b : SUCCESS in 12m 54s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a2293d73c5ab4d4b99f8807a13859ef2 : SUCCESS in 18m 13s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6225b631bf9442a49ff2b6a4f4c693dc : SUCCESS in 14m 08s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d43a2189252344e881fd337e236fd3a9 : SUCCESS in 13m 36s (non-voting)\n- puppet-placement-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f59c1fcc7d1c40058689982a40a6bb45 : SUCCESS in 1h 50m 47s","accounts_in_message":[],"_revision_number":1},{"id":"963802aba6e08c9fce1e467a01f082b840985ad0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-29 08:18:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"718a3217bdd3d69bf0283df95d6933945fc891b0","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-11-29 11:41:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"47aca95a20a255377a5ad10b130c3f10aa865f18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 11:43:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa998851c3f32ddbd74039623177c5d0c3709003","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 13:52:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/735fb0af93c941c2bbf657c9125f6e28 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a07fa1f15dcf47d88c57596e586c3366 : SUCCESS in 3m 48s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e43302b38655425b99909575cfdcdb48 : SUCCESS in 3m 41s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c363e5d371447ee9f4f0ccd7eba7a48 : SUCCESS in 11m 36s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9ee4ba68c28c41a3abfc27f83ed11563 : SUCCESS in 12m 22s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6fad5e0c593460995915a23de5a6c60 : SUCCESS in 8m 17s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/700b2e50de80469ca35755e7dd7959e0 : SUCCESS in 8m 47s\n- puppet-openstack-integration-7-scenario001-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9988a32aae094db9aee03a0ea7d259e9 : SUCCESS in 1h 09m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9d7f1676f7cd43b38e3b29edfacb5407 : SUCCESS in 1h 18m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/454164f154d4432e8f215b39799d60fa : SUCCESS in 1h 30m 30s\n- puppet-openstack-integration-7-scenario004-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b6fc0f2122e468a9c5bbc96866c8763 : SUCCESS in 1h 20m 56s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b1f0fb5d956245ec8bbc2caa6ab3de28 : SUCCESS in 18m 58s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/611af634eaf24cdf9e70e8962f8e5cc4 : SUCCESS in 13m 53s (non-voting)\n- puppet-placement-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a47ea80b6eb648418f5c108d1b8610e7 : SUCCESS in 2h 01m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c3ae05ca27fce4590e84ce5a887f9bc17fe2de36","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 13:52:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8a42cefcb32de0bedee3fdddab21bfd16685736d","revisions":{"8a42cefcb32de0bedee3fdddab21bfd16685736d":{"kind":"REWORK","_number":1,"created":"2021-11-29 00:57:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/35/819535/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/35/819535/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/35/819535/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/35/819535/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/35/819535/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/35/819535/1"}}},"commit":{"parents":[{"commit":"0409934acec3735b8ca4e413a85a1e01a5c9a603","subject":"Use consistent format for service description","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/0409934acec3735b8ca4e413a85a1e01a5c9a603"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 00:57:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 00:57:45.000000000","tz":540},"subject":"Fix dependencies related to openstacklib::policy","message":"Fix dependencies related to openstacklib::policy\n\nSince [1] was merged, not only openstacklib::poliy::base but also\nopenstacklib::policy::default is included to manage the policy file.\nThis change ensure openstacklib::policy::default is executed after\nthe packages are installed.\n\n[1] a4bd6e6c877aa6a9ccf71eb2c8cd9ef2e6ffdcd5\n\nChange-Id: Ie9c8fa687ca57d7bc49d6f35941efbc94a3f5344\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/8a42cefcb32de0bedee3fdddab21bfd16685736d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/8a42cefcb32de0bedee3fdddab21bfd16685736d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
