)]}'
{"id":"openstack%2Fpuppet-placement~854761","triplet_id":"openstack%2Fpuppet-placement~master~Ia5cff022c1ad4f6402f2e9d4cd039059b176cd93","project":"openstack/puppet-placement","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-08-26 18:09:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia5cff022c1ad4f6402f2e9d4cd039059b176cd93","subject":"apache+mod_wsgi: Allow customizing wsgi process options","status":"MERGED","created":"2022-08-26 08:50:58.000000000","updated":"2022-08-26 18:12:00.000000000","submitted":"2022-08-26 18:09:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"854761","meta_rev_id":"df50cb4ca4be313c6ce060503ce024d691cbd717","_number":854761,"virtual_id_number":854761,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-26 18:09:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2022-08-26 11:17:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2022-08-26 16:02:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2022-08-26 16:02:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-26 10:35:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-26 11:17:31.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2022-08-26 16:02:40.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"720f1f4b203619e85fe1bd1b7dfede9b376e01fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 08:50:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c24c1bf4965f700e3d4b450c6e920a2861ce6f5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 08:55:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"82d3ee481053000fcffba1d54d81f31b995bf307","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 10:35:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/817e69a6d3264de5b689c6aaf30ce585\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb3e7f6337bb46b1941e727720ad8ea7 : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6907fedd25c84b8d98b0b4ef49dbc213 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2fc2aa441a44448b4f3e00e10a4ed49 : SUCCESS in 3m 25s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/de42ea1ba5a147809a0bb378d7ca9543 : SUCCESS in 17m 51s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fd1c69e52da4f6db9fc58e0901694ff : SUCCESS in 10m 16s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/be67b3bccb7c4b5cad1a5182518ea4e2 : SUCCESS in 10m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba4be2fab541437181f5dcf463c0d9e3 : SUCCESS in 11m 00s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d9b7bd34e3e4d9fbb405e8f0c8c0486 : FAILURE in 46m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afc5e8be3a8d413cb3cee1f6d628d693 : FAILURE in 55m 29s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/816e1e62dffb4ea7bc81b19d6e3c0e22 : FAILURE in 44m 50s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89e86dae8a6145d2a8ad30fd5e8884b6 : FAILURE in 34m 48s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/563b7ba8517340cc8ba28a72672ce7bb : FAILURE in 39m 46s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b32953651be24375b106cd35d64205e7 : SUCCESS in 24m 55s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d28558fde5ed4d9eb891ce2e0f4e1ae5 : SUCCESS in 1h 00m 45s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e8cedaac4ea4ce58c56994c98ea60c1 : SUCCESS in 1h 05m 00s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7a03c63bd0634ec6a8652dd1af168f2d : SUCCESS in 1h 09m 57s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/f22de2825af24883b6e20b68ef04f340 : FAILURE in 56m 28s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f96d538cc4d49fe89fb11c91f47db0d : SUCCESS in 50m 13s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd1f650bb643440389f938b76e76ba60 : SUCCESS in 17m 25s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/bd02e2b79e8d485aa25ac82393a1fa7a : SUCCESS in 24m 08s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/786a1de755db4b789f0e171ab45b7f62 : SUCCESS in 16m 01s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e2b4af8c0d74d13be17ba42b20020f9 : SUCCESS in 15m 57s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0f075fe9fd846be9570604d3998fe6e : SUCCESS in 4m 00s\n- puppet-placement-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3d07fddde74048c790c9afba739b2655 : SUCCESS in 1h 36m 39s","accounts_in_message":[],"_revision_number":2},{"id":"d1230fb0c6148fca0ab710fac337cb8e2d94faca","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-08-26 10:46:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b03ccb263ed715db0d6ad452c68b46613f5a4729","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2022-08-26 11:17:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"254cd3cc9d0e94e5d75225cac8467e87f7937389","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 12:19:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25157ba35e654c8da7d6bbac3e2d6f30\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/369b449686ee43e4a1c398175a6b8cb0 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b0736c0c6af4489aaf86da16f72bc3a : SUCCESS in 3m 32s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23c0577d4abc4a988eb3d00e2f8761ef : SUCCESS in 3m 37s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/9858335d325642b0b293ed2bcdd15bf1 : SUCCESS in 16m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b83a25294b6d498c84e0b03dc2d8f32e : SUCCESS in 10m 39s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/492477a5ffb741b8aa91d968fb1fac2d : SUCCESS in 11m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d40889c217634defbc70e1c63ea908c9 : SUCCESS in 10m 37s (non-voting)\n- puppet-openstack-integration-6-scenario001-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0921b960bd524aa09bb90c8b1bfde02c : FAILURE in 52m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87d2fc7c388446759e7ca0a9b411f2de : FAILURE in 1h 01m 31s (non-voting)\n- puppet-openstack-integration-6-scenario003-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f116a2cca26420baad9e1044f59cf13 : FAILURE in 50m 13s (non-voting)\n- puppet-openstack-integration-6-scenario004-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39a58708fcb04a7ab1e5f62cf6e09e82 : FAILURE in 36m 53s (non-voting)\n- puppet-openstack-integration-6-scenario005-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ca431b27ccb4261891b09d3bcdf1deb : FAILURE in 39m 16s (non-voting)\n- puppet-openstack-integration-7-scenario000-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/b611d83ce28643b4ad20995cfd816094 : SUCCESS in 24m 23s (non-voting)\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5180a0b65e545c699155e079897dad8 : SUCCESS in 1h 03m 04s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f030526a97c4226abc14bb9fc3e0b56 : SUCCESS in 1h 19m 41s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c216e512e8364d3a99047f694ea3f6c7 : SUCCESS in 1h 06m 27s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/0003bab3468f47789c796931e8fde279 : SUCCESS in 1h 01m 34s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4885631aa6894da39d235932d0fb5ad5 : SUCCESS in 53m 30s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6132a3c38a144ecbc4727c1fe3efd4c : SUCCESS in 16m 14s (non-voting)\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/971d88af0aa14f26b022c174ea1dda1c : SUCCESS in 23m 39s\n- puppet-openstack-litmus-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5803eb51c33347d5956bfc4bbd409a94 : SUCCESS in 14m 26s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66a5ca08188b4ea6a487b07350ff1421 : SUCCESS in 24m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d12f02f871224a0f97f4e1d1a836fd06 : SUCCESS in 3m 25s\n- puppet-placement-tripleo-standalone https://zuul.opendev.org/t/openstack/build/53cedc1a0c1e41cabc4fae5d525bcb56 : SUCCESS in 1h 29m 32s","accounts_in_message":[],"_revision_number":2},{"id":"70c4666c9b2eab3bfb6627dfb8b9c1f8530240be","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2022-08-26 16:02:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f414335b3b67b21bcd325a5e9adb45d10c79e436","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 16:04:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0a3861fb366c191d9f0aba9c12a5a4c7a50bf4eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 18:09:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26e56938ce1e4fb09bf65ba4000afcbd\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25dec7ba3f354ea39df0064276c6f110 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b2fc56f67ce41e1bcab485a0f4a1fd8 : SUCCESS in 3m 44s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5a29f949b9f42e0a04d6b538f134ef5 : SUCCESS in 3m 43s\n- puppet-openstack-unit-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/969b54d2c8e7443d8b3f1b840eb6ce68 : SUCCESS in 13m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a37bdf1504c41518065e7bc31c9a971 : SUCCESS in 7m 41s\n- puppet-openstack-unit-7.16-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b97c1799b2fc461f8edcba5a75a3a97f : SUCCESS in 8m 06s\n- puppet-openstack-integration-7-scenario001-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c90b8c19967f47ee98fdf800bd5866ea : SUCCESS in 1h 06m 30s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1faf16a266a48f9bac48a13dc731905 : SUCCESS in 1h 15m 55s\n- puppet-openstack-integration-7-scenario003-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/a53733802e7a4a92b1885a8a32ab3c0d : SUCCESS in 1h 08m 13s\n- puppet-openstack-integration-7-scenario004-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/abee84878e2541498c3eb7009150d6c0 : SUCCESS in 55m 48s\n- puppet-openstack-integration-7-scenario005-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1593afef48234b1696b0910c7e3a4041 : SUCCESS in 57m 11s\n- puppet-openstack-litmus-7.16-centos-9-stream https://zuul.opendev.org/t/openstack/build/8aa912a692dd4f4c9f861064b8ef56d0 : SUCCESS in 23m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1969409ab11f412db50cea090c50694d : SUCCESS in 4m 12s\n- puppet-placement-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2b7c032e4a6340f581907f7a0ad5286d : SUCCESS in 1h 41m 37s","accounts_in_message":[],"_revision_number":2},{"id":"4959f5c97ec0ec39abd2a59c67be3b0f28d5468c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 18:09:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"df50cb4ca4be313c6ce060503ce024d691cbd717","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-26 18:12:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2414906c0c3f40fdad5b3e33e2873241\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fff366e191eb4a878e8126f1a2c0964b : SUCCESS in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a945887286c9cfe4f9dcfb081424236e4d094788","revisions":{"f632fb3f58eaed5b146f1aa15790f90817e4404c":{"kind":"REWORK","_number":1,"created":"2022-08-26 08:50:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/854761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/61/854761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/61/854761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/61/854761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/61/854761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/61/854761/1"}}},"commit":{"parents":[{"commit":"497d0a4e457746215ed9bd2f32738fd1707c7796","subject":"Add Apache WSGI logging parameters for pipe/syslog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/497d0a4e457746215ed9bd2f32738fd1707c7796"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 08:50:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 08:50:19.000000000","tz":540},"subject":"apache+mod_wsgi: Allow customizing wsgi process options","message":"apache+mod_wsgi: Allow customizing wsgi process options\n\nThis change adds the missing capability to customize wsgi process\noptions, which is supported by the other modules.\n\nChange-Id: Ia5cff022c1ad4f6402f2e9d4cd039059b176cd93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/f632fb3f58eaed5b146f1aa15790f90817e4404c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/f632fb3f58eaed5b146f1aa15790f90817e4404c"}]},"branch":"refs/heads/master"},"a945887286c9cfe4f9dcfb081424236e4d094788":{"kind":"REWORK","_number":2,"created":"2022-08-26 08:55:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/61/854761/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/61/854761/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/61/854761/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/61/854761/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/61/854761/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/61/854761/2"}}},"commit":{"parents":[{"commit":"497d0a4e457746215ed9bd2f32738fd1707c7796","subject":"Add Apache WSGI logging parameters for pipe/syslog","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/497d0a4e457746215ed9bd2f32738fd1707c7796"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 08:50:19.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-08-26 08:55:37.000000000","tz":540},"subject":"apache+mod_wsgi: Allow customizing wsgi process options","message":"apache+mod_wsgi: Allow customizing wsgi process options\n\nThis change adds the missing capability to customize wsgi process\noptions, which is supported by the other modules.\n\nChange-Id: Ia5cff022c1ad4f6402f2e9d4cd039059b176cd93\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/a945887286c9cfe4f9dcfb081424236e4d094788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/a945887286c9cfe4f9dcfb081424236e4d094788"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[]}
