)]}'
{"id":"openstack%2Fpuppet-placement~947399","triplet_id":"openstack%2Fpuppet-placement~stable%2F2025.1~Iba6ec55c4a18e3d80fb288ef2295496321d3c368","project":"openstack/puppet-placement","branch":"stable/2025.1","topic":"create-2025.1","hashtags":[],"change_id":"Iba6ec55c4a18e3d80fb288ef2295496321d3c368","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","status":"MERGED","created":"2025-04-16 08:41:59.000000000","updated":"2025-04-16 09:40:41.000000000","submitted":"2025-04-16 09:39:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947399-create-2025.1","meta_rev_id":"c6a5bcfdbc54500ceaa106451aa93916e444aa88","_number":947399,"virtual_id_number":947399,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-16 09:39:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2025-04-16 09:26:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2025-04-16 09:26:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-16 09:22:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-16 09:26:53.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"b6f2c0cbb74865a03f6c3f9cdaecef0bb574a30e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-04-16 08:41:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63884d4d1e6803ba74a205d72c977ba6b2505ec5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 09:22:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/765867b6a033409ba9e63a6da1589be1\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/49f86df778fc4f1faf1449d58ceac424 : SUCCESS in 2m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcf17c6bc0ad4014a70a311615bd4bb7 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"d12e3534df195a92ac2f87ef26ca4ee924280c2e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-04-16 09:26:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"25dab91001ef7a5f9f885152615550e5d56e3b45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 09:35:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c992fdd1485cddaaca56e335f52aa86e12fa019c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 09:39:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f8ac8a128e3473f81ba7554c5ee5dd7\n\n- puppet-openstack-lint-ubuntu-noble https://zuul.opendev.org/t/openstack/build/76a2486ca43c4b948b59da175c045792 : SUCCESS in 2m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9777cb237ec840acb44132cb2c3e11aa : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":1},{"id":"15b42e226afedc9b2fc11338301bb5390e1683cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 09:39:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c6a5bcfdbc54500ceaa106451aa93916e444aa88","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 09:40:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04e4067d8a21427c885fb885a20b44c8\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87076aadc5e540e6985295570537a1fa : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"535f942678333ff9a6a4c1b4a59f23773b923201","revisions":{"535f942678333ff9a6a4c1b4a59f23773b923201":{"kind":"REWORK","_number":1,"created":"2025-04-16 08:41:59.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/99/947399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-placement","ref":"refs/changes/99/947399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/99/947399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/99/947399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-placement refs/changes/99/947399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-placement refs/changes/99/947399/1"}}},"commit":{"parents":[{"commit":"c4a25a8940220e6abda9be5724df57eef61b2459","subject":"Update .gitreview for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/c4a25a8940220e6abda9be5724df57eef61b2459"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-04-16 08:41:57.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-04-16 08:41:57.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","message":"Update TOX_CONSTRAINTS_FILE for stable/2025.1\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/2025.1 branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Iba6ec55c4a18e3d80fb288ef2295496321d3c368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/535f942678333ff9a6a4c1b4a59f23773b923201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-placement/commit/535f942678333ff9a6a4c1b4a59f23773b923201"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
